)]}'
{"id":"openstack%2Fpuppet-vitrage~660001","triplet_id":"openstack%2Fpuppet-vitrage~master~I1e5d1140312684c993da06e9d14f6d287e691f06","project":"openstack/puppet-vitrage","branch":"master","topic":"fix-client-test","hashtags":[],"change_id":"I1e5d1140312684c993da06e9d14f6d287e691f06","subject":"Fix client spec test","status":"MERGED","created":"2019-05-20 03:03:58.000000000","updated":"2019-05-20 04:33:46.000000000","submitted":"2019-05-20 04:33:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660001-1558326826618-795c03e7","meta_rev_id":"e1c03acf9abed280b4b96e0775d9ff539d9c1b35","_number":660001,"virtual_id_number":660001,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-20 04:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-20 04:17:50.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 04:17:50.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 04:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd6e296db505f04388e355be37f62bbf306a815b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 03:03:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39cba7ee98cebe3495cca4efc1c552c100a6ea0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:23:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/01/660001/1/check/puppet-openstack-lint-ubuntu-bionic/0e43b92/ : SUCCESS in 6m 25s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/01/660001/1/check/puppet-openstack-syntax-4-ubuntu-bionic/a0042c7/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/01/660001/1/check/puppet-openstack-syntax-5-ubuntu-bionic/5aac2c2/ : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/01/660001/1/check/puppet-openstack-syntax-6-ubuntu-bionic/714bb93/ : SUCCESS in 3m 23s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/660001/1/check/puppet-openstack-unit-4.8-centos-7/b36a094/ : SUCCESS in 13m 54s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/660001/1/check/puppet-openstack-unit-5.5-centos-7/f50c2c7/ : SUCCESS in 10m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/01/660001/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/f4d696f/ : SUCCESS in 6m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/01/660001/1/check/puppet-openstack-unit-latest-ubuntu-bionic/ef0435d/ : SUCCESS in 7m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"252d716001cc218cb1ba24f29c00a161c5d16e05","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 04:17:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3750599d3184a9ca2a66a50ea9d670b85a48f6ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:18:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f55852d3b3ce028fed4fc35317b2b7ab3f89f8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:33:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"666bd591adbd9b45c493e9fb17085767563ebe95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:33:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/01/660001/1/gate/puppet-openstack-lint-ubuntu-bionic/9e2f611/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/01/660001/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/400a53d/ : SUCCESS in 3m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/01/660001/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/40355e3/ : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/01/660001/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/a78dbae/ : SUCCESS in 4m 18s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/01/660001/1/gate/puppet-openstack-unit-4.8-centos-7/f5ff5ba/ : SUCCESS in 13m 36s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/01/660001/1/gate/puppet-openstack-unit-5.5-centos-7/b2d64b3/ : SUCCESS in 10m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/01/660001/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/fe85248/ : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"416218e03d9b3a0bdb63ec070c5e6ad1cd3e304a","revisions":{"416218e03d9b3a0bdb63ec070c5e6ad1cd3e304a":{"kind":"REWORK","_number":1,"created":"2019-05-20 03:03:58.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/01/660001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-vitrage","ref":"refs/changes/01/660001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/01/660001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/01/660001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/01/660001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-vitrage refs/changes/01/660001/1"}}},"commit":{"parents":[{"commit":"a5010d2c4704a50c5c793a2d231c3f2fd400013d","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/a5010d2c4704a50c5c793a2d231c3f2fd400013d"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:21.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:21.000000000","tz":480},"subject":"Fix client spec test","message":"Fix client spec test\n\nChange-Id: I1e5d1140312684c993da06e9d14f6d287e691f06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/416218e03d9b3a0bdb63ec070c5e6ad1cd3e304a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/416218e03d9b3a0bdb63ec070c5e6ad1cd3e304a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
