)]}'
{"id":"openstack%2Fpuppet-vitrage~724046","triplet_id":"openstack%2Fpuppet-vitrage~stable%2Fussuri~Ieca3a71c4dd00330c4e3fcc695e27a74e078f424","project":"openstack/puppet-vitrage","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ieca3a71c4dd00330c4e3fcc695e27a74e078f424","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-28 15:00:32.000000000","updated":"2020-04-29 02:54:21.000000000","submitted":"2020-04-29 02:54:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724046-1588128861398-e63aa4fb","meta_rev_id":"10185d8d8d9791ae54d684b8ad7d8b8edb35e4f8","_number":724046,"virtual_id_number":724046,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 02:54:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 01:58:28.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-28 16:48:43.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 01:58:28.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:48:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 01:58:28.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-29 02:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2097ca8b91d73124d49545b66c52237c246b463","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 15:00:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9801a1c57a4e23d6ad29183fc03cf1a7b220b13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:32:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c7e1d6c84924775943d89fc924ffb55 : SUCCESS in 7m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd9b4682dcb8416c9317f35a2a2849e9 : SUCCESS in 7m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffae9cded1d74609b16d134893a5f2e5 : SUCCESS in 6m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ad3151b42ec94ed090f090c2f99f4d4e : SUCCESS in 19m 14s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a40a2bfd49734d4cb35cd5389d43e927 : SUCCESS in 15m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d5e190c463d41fbaddffa09e3c02b80 : SUCCESS in 13m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e720f81381046e6bc5831a5645ab152 : SUCCESS in 13m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/390f07d0842847db92ca619e67e6358a : SUCCESS in 47m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/31fec5fd0c394a1e8b1d51902ec74356 : SUCCESS in 1h 01m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1a9520542eb64302bf0022e886164128 : SUCCESS in 26m 35s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/45921a14816f4b878530c66049c10e76 : FAILURE in 13m 37s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ef660cc27a3492494e593dcf3984003 : FAILURE in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2dcba238ee9b9ca54480ae51ea0f0b236fbfa0ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:48:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"787da191bbdddea477f9a99af6f9404a9a60d1ab","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 01:58:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fd30006f566e6bc48d70b3c745e63170f9245f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:58:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adb15588364422c7cfa079f7736f601af994ec38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:54:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88a6b6a0b3ed933f4de3eabc1e0e90f9b4617d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:54:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88f38f30973149e184551415031f7361 : SUCCESS in 6m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e7789b8c6384acf9dff4668050ffe88 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7e5251f895d4f4e9af40f9589272ccb : SUCCESS in 7m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a61b9cd3ab6d407d8a91ae320766b686 : SUCCESS in 12m 40s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1cf7d5ff7f3946dca6db8669f5d58c10 : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98547d526d294fc4802dca540e3d24db : SUCCESS in 8m 53s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b34fde41016246b8833806bab7625ccf : SUCCESS in 47m 27s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/18147d1f75e34bd49c9035b7af25e4e4 : SUCCESS in 53m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/93c6ae33296e4f2b9e80e45d37e669bb : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e9c68f2d90eec68b2b7f122a6ebf2823783a254","revisions":{"1e9c68f2d90eec68b2b7f122a6ebf2823783a254":{"kind":"REWORK","_number":1,"created":"2020-04-28 15:00:32.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/46/724046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-vitrage","ref":"refs/changes/46/724046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/46/724046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/46/724046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/46/724046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-vitrage refs/changes/46/724046/1"}}},"commit":{"parents":[{"commit":"7aa493e950cc187ac6bdd9fd2ed3efaa5d92fa88","subject":"Prepare Ussuri RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/7aa493e950cc187ac6bdd9fd2ed3efaa5d92fa88"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 15:00:30.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 15:00:30.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Ieca3a71c4dd00330c4e3fcc695e27a74e078f424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/1e9c68f2d90eec68b2b7f122a6ebf2823783a254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/1e9c68f2d90eec68b2b7f122a6ebf2823783a254"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
