)]}'
{"id":"openstack%2Fpuppet-vitrage~947463","triplet_id":"openstack%2Fpuppet-vitrage~master~I7d02f9344aebc672c2262894f76a1f49177a3f6d","project":"openstack/puppet-vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-17 08:54:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I7d02f9344aebc672c2262894f76a1f49177a3f6d","subject":"Remove watch_log_file","status":"MERGED","created":"2025-04-16 12:00:39.000000000","updated":"2025-04-17 09:57:56.000000000","submitted":"2025-04-17 09:57:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947463","meta_rev_id":"f6d838a57733adea7b7c17a878dd96d26a8876fd","_number":947463,"virtual_id_number":947463,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-17 09:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-04-17 08:54:14.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-04-17 08:54:14.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 13:13:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-17 06:38:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"}],"messages":[{"id":"7781da812f6f72599c548b8523e36e3169f28107","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-16 12:00:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5362a56c8a6f4c926ae4a243cddc519a184669d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 13:13:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba13a501d87d4d55a16fae9c7312814b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4279085adf94a54ad2d59d70349c135 : SUCCESS in 2m 14s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ebd414c3a34386a55c2337a0678c1a : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bb84569bf8245b1bf311a39f7dfcb06 : SUCCESS in 2m 48s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/36de930b87e641cd896f4866902daefc : SUCCESS in 4m 02s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62ccb0bb880e4b7fa4e6e38cf6eaa4d5 : SUCCESS in 4m 28s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abfdee15499543cdb1da0aa11ba3aca0 : SUCCESS in 5m 13s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee7ad2e706f14cacadd904f89f5ba570 : SUCCESS in 24m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43e5e3bef6f240e39192f76c5705dc9e : POST_FAILURE in 1h 07m 33s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ad853140dcc4c6a87bb12facde9755e : SUCCESS in 20m 50s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d37b8f468da48e188cfd33c7750d950 : SUCCESS in 10m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2ff558fe71f4776b536b14aedd5642a : SUCCESS in 11m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aacfb7b763149ecba81e13722f12874 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f4cf21a28da07f8e897866df99ace3bf304d99b3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-17 00:56:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7920afed92263bfad2a284c5ef47bb6bf5302e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 02:15:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e9c0eb764dd40e6a52b21c1828c1122\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a55b41042db4413ae7dc0d7d5397df2 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2da47921e0d43a280395853e58f0eb5 : SUCCESS in 1m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e75cd9884fe94209b7fd0a5ffc4ac7c7 : SUCCESS in 1m 43s\n- puppet-openstack-unit-7.35-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8cbea77d2064cc092051a2f4da62180 : FAILURE in 3m 19s\n- puppet-openstack-unit-8.11-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19f3342fafc0478eb7f216eb1db775c5 : FAILURE in 2m 28s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15741deeb36b4aea8cdb8158934edbd5 : SUCCESS in 4m 32s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/914cd6cacbf24de794df35abb5ddd285 : POST_FAILURE in 1h 04m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ade3fb55998472b8357afb18d602f08 : SUCCESS in 1h 13m 06s\n- puppet-openstack-litmus-7.35-centos-9-stream https://zuul.opendev.org/t/openstack/build/2acd761f2096450a8dc5f9f8d21bb3ec : FAILURE in 5m 12s\n- puppet-openstack-litmus-8.11-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad44a631f97c433c858792430d1547fd : FAILURE in 2m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/133c31ee831d49dcb994eee5ac3fcd20 : SUCCESS in 12m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9245792b6304b02bfbd1b32a3a89fc8 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":1},{"id":"caa7337ad0b0961a0846684156ab92db272a81b2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-17 06:38:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bedd7e92aa2092fa9c8decc69fea92695f78b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 07:36:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50bdf3bf3dd244c5ac5e7a2a902dc437\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7624b0ec4e144c8aa91c1bb372d67b44 : SUCCESS in 1m 19s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9eabeb341324ab585067f73ef15e43f : SUCCESS in 2m 11s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/593f9fd50fa64c6aabc3398c322c43ad : SUCCESS in 3m 08s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9167b065b0c4b78915391ce42b6f1cc : SUCCESS in 8m 01s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34f17015872e4e8bb9dd1b65314159a2 : SUCCESS in 3m 51s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b7bf53f63ad4e0c8642cbfeb7f122e2 : SUCCESS in 3m 06s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6df7d5b5fcf945b0a81ccf19fab18c37 : SUCCESS in 52m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2b7ec5d21b44557be79d82f2a19ed66 : SUCCESS in 52m 55s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/41cd1da7da34412ab86888da518f0ca3 : SUCCESS in 20m 10s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf88cd85c36d4f31b4dff1a8640797ca : SUCCESS in 6m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06a6ba56efda418c8e6dc0fe9a855b25 : SUCCESS in 11m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30d3c5cd152b41d4b35982302cccc41f : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2b922aa0dbec367fcb937c1cc9faa2037b6dbfb3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-17 08:54:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7ba32de81e2200331efe2a41765ff01ef2f806d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 08:55:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc9f01a3ec9cef94537ef287e118044593010ca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:57:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfae90aa9c72403aaf62a16d363f2727\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91ff8f5eee5a439483394b7e87c13b37 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32a0b4f6f6a34b34831a6e8c0dfe5093 : SUCCESS in 2m 17s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9745dd6e9904f1f88b851e79bd161c4 : SUCCESS in 2m 09s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/060b8fc61e9942ecabdde39df809db71 : SUCCESS in 8m 49s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e3e5d39ee224678b868cdb9a2ae155f : SUCCESS in 4m 42s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61b6e72721bd423897df60cbc83e404d : SUCCESS in 52m 42s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/465b90dc55624791ad2d116a63199178 : SUCCESS in 19m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bbf71acd8244691af2fb2ca2fb49b03 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a4a370e56744714c83772c42cec87e834e1771d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:57:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f6d838a57733adea7b7c17a878dd96d26a8876fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:57:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70741b854e504918bcabeae4d57cc37c\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/242f54c89ffc48308bffbe320a4c0d65 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efdb2dc2ad6b4c014bfa55acdabcaed82fd66209","revisions":{"efdb2dc2ad6b4c014bfa55acdabcaed82fd66209":{"kind":"REWORK","_number":1,"created":"2025-04-16 12:00:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/947463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-vitrage","ref":"refs/changes/63/947463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/63/947463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/63/947463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/63/947463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-vitrage refs/changes/63/947463/1"}}},"commit":{"parents":[{"commit":"dd65deea5523f89b03d7c85d553c129f1d69f999","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/dd65deea5523f89b03d7c85d553c129f1d69f999"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 12:00:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 12:00:31.000000000","tz":540},"subject":"Remove watch_log_file","message":"Remove watch_log_file\n\nThe option was deprecated during the 2024.2 cycle[1]. Processes may\nfail to start when it is set to True[2] so supporting it is no longer\nuseful.\n\n[1] 31603ac6dacd0b47ec9f02d99bfdd0a7b683119c\n[2] b977cb9becbeaaa67bb663bd8db5aa6846e495b9\n\nChange-Id: I7d02f9344aebc672c2262894f76a1f49177a3f6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/efdb2dc2ad6b4c014bfa55acdabcaed82fd66209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/efdb2dc2ad6b4c014bfa55acdabcaed82fd66209"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
