)]}'
{"id":"openstack%2Fpuppet-vitrage~964986","triplet_id":"openstack%2Fpuppet-vitrage~stable%2F2025.2~I4fae3ddcc6f0fb143fe64243e8290c9eb914fc3d","project":"openstack/puppet-vitrage","branch":"stable/2025.2","topic":"create-2025.2","hashtags":[],"change_id":"I4fae3ddcc6f0fb143fe64243e8290c9eb914fc3d","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","status":"MERGED","created":"2025-10-27 17:07:06.000000000","updated":"2025-10-28 01:40:31.000000000","submitted":"2025-10-28 01:39:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964986-create-2025.2","meta_rev_id":"a8c9e2cc00084ad6b6f382e22f82008ddaf008f1","_number":964986,"virtual_id_number":964986,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 01:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-10-28 01:24:41.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-10-28 01:24:41.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 17:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-28 01:24:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"8986001e2b2aa196acee2a3323b4e2b293c8767d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-10-27 17:07:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e5328d3960db709584d407da918d440b89d546c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:21:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220368ede7034a0c9e4cc32c5cf2679b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ace27c3cc554e7b84bebaec0843ee90 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45ca9200ac3542cc90c8616287ab95f0 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3996923f9ea3a1b3a16a5732fdf742ff9669ef76","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-28 01:24:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bd9725052b15a8be5879bd49472cfdcdd7a49d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 01:25:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f37c262990e95e6b69eab52e92a366cd6f5ab2d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 01:39:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d81cfb6603354d91bf3bb5e6848494c9\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ec8dfb8946745ad93d3454a673d4c48 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7992976dbe184b11ae52b5902bacd876 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3e8bab54374f1d7ac5c30f5a7823d63407cf709a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 01:39:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8c9e2cc00084ad6b6f382e22f82008ddaf008f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 01:40:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79975d1a998944df942d1288b58e9b27\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b3bcbc53a9e499db246d016093a60c3 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"918f63f9e819aa30d09ca8eab0dcaf9820e2714e","revisions":{"918f63f9e819aa30d09ca8eab0dcaf9820e2714e":{"kind":"REWORK","_number":1,"created":"2025-10-27 17:07:06.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/86/964986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-vitrage","ref":"refs/changes/86/964986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/86/964986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/86/964986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-vitrage refs/changes/86/964986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-vitrage refs/changes/86/964986/1"}}},"commit":{"parents":[{"commit":"a5b7a65d0f6ac27c91dd579a0c9bb59e04d05d6d","subject":"Update .gitreview for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/a5b7a65d0f6ac27c91dd579a0c9bb59e04d05d6d"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-10-27 17:07:05.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-10-27 17:07:05.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2025.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2025.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I4fae3ddcc6f0fb143fe64243e8290c9eb914fc3d\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/918f63f9e819aa30d09ca8eab0dcaf9820e2714e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-vitrage/commit/918f63f9e819aa30d09ca8eab0dcaf9820e2714e"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
