)]}'
{"id":"openstack%2Fpuppet-zaqar~796239","triplet_id":"openstack%2Fpuppet-zaqar~master~I35b86a599d9e947648cad38904d54a8f998696e1","project":"openstack/puppet-zaqar","branch":"master","topic":"pyvers-removal","hashtags":[],"change_id":"I35b86a599d9e947648cad38904d54a8f998696e1","subject":"Get rid of the $pyvers variable","status":"MERGED","created":"2021-06-14 07:45:03.000000000","updated":"2021-06-15 23:07:28.000000000","submitted":"2021-06-15 23:07:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"796239-pyvers-removal","meta_rev_id":"7a24d43102398fe3a032fd340f304778a2b1e0b5","_number":796239,"virtual_id_number":796239,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 23:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-15 11:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-15 16:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-15 16:39:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 09:26:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 10:32:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-15 16:39:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"72f9275bb1bf40ca771646c6cd587c7f03139d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-14 07:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c111d5cc5232526b31fab4ab59d5427b06a7cab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 09:26:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b138e8222be46d68bb397565263abad : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cf7266f164a4778a14a41a8aba9f3ff : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca4582d5f44743a4b24bae2e37288a65 : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf9dc47ab96f44cca374c4302ee57a59 : SUCCESS in 11m 39s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/5268d932db5f49aba5af20cedd6a5bc5 : SUCCESS in 12m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b1349c1ef994afaa273e972205f7100 : SUCCESS in 10m 37s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fae6f7dc5dc4499a82af760877d4f16a : SUCCESS in 8m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f46136e8442b4937828fb9ccb8d088a2 : SUCCESS in 8m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26bee4c774964b71a6d61b79df237eb8 : FAILURE in 1h 16m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/21b1ebfd058c4af587bf98b32d205d75 : SUCCESS in 1h 38m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f67026c15430469eaf705c6b5f69c28b : SUCCESS in 1h 29m 15s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/18da756f997c4f91936038aeb7115946 : SUCCESS in 29m 48s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ce53eb46cca4dfa9f00d8166e4003a2 : FAILURE in 21m 53s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a1c9df634904df39e0e0954dd11435d : SUCCESS in 25m 27s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c5fea8803b84fdeb4d77704f3fb5229 : FAILURE in 16m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b244af7d40e24f4e9c7af1b48e15ed44 : FAILURE in 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c003124ff33d1f60b9a541c5870460fa92c362f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-14 10:32:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29d635fd713c83dd2ce753cb9d1e6150b4b49ae3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-14 10:39:09.000000000","message":"Patch Set 1:\n\nSee https://review.opendev.org/c/openstack/puppet-zaqar/+/796130 for your reference.","accounts_in_message":[],"_revision_number":1},{"id":"837e8f24943f7d64d3bfed9243fbc54e8ae6aa75","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-14 11:39:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b9457ea03b62427e7002d74bfdf259689398919","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-14 12:30:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a488e8f8e6a77966a0181c66137703e658ab958d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-15 10:59:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf85ac715f8bd762220882abe9bf421a69238897","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 11:33:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c53d6ffce88e389ce83368f24224ec28ce2f7153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 12:40:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d238103291e04a4a824dc2df3256cf97 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6720956413a6465781d2858ad7811880 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/656b8ef326d24d73b49b9b4631a6822f : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/73412fc0eb3e47afbd4bfb4beebc2629 : SUCCESS in 12m 25s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8efe14b2dff145b5b7828738faed80ac : SUCCESS in 10m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21246a0eeb104ebd84df39902bfd969e : SUCCESS in 9m 52s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa3cf16e1b3940788ef914f17a98fef0 : SUCCESS in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c13aefcafde04aa78ee3474f71cf0411 : SUCCESS in 9m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89be7a16b8b149588c54205b1ca63b3f : FAILURE in 44m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a334fb3b6af3456eaf73cd5ecaec42c4 : SUCCESS in 1h 28m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/142d785b9830412ca2f9864a0d411931 : SUCCESS in 1h 37m 47s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/165f419d40654b9db9124d65dd593bef : SUCCESS in 23m 56s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c194f6b692b446ab9ec2092a15c10190 : FAILURE in 20m 29s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1bb27625c0b44d148e14a8b086c165d7 : SUCCESS in 27m 03s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6c49fd467754dfaab327aca1b2ecb76 : FAILURE in 17m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a41e8de9215b4c97b7991a63e19c79d7 : FAILURE in 22m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"619d8eff79bc29c7d52c1bfa69a8dead8c77e7d5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-15 16:39:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"02d4ee97cdae06f3e2be71288f99d1dd8b007b09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:39:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"160e3f965cd5e9448848a627846cebd00752438a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 18:45:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/043fc98e2dca40908d648701543028ef : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf5357b2883d4e789f12f8a558962255 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0208fb5896be4f8cb1c2aa1ef01f2341 : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/619dd643c32540fd8d70e84ce98c9aa9 : SUCCESS in 11m 10s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0071feb7547411ba86ebf36862cb653 : SUCCESS in 11m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/013103dae24649c4a4e7974a1b5d8881 : SUCCESS in 8m 12s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b32f3bbc0e2f4bae80184d4bf4551276 : SUCCESS in 8m 20s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/25700f474d4c41528330244f0d3d172c : TIMED_OUT in 2h 03m 00s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/34b229feb11846b091e16cb3b58da692 : SUCCESS in 24m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/799a06c71f5e46079010c181b71f824e : FAILURE in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a29d9009343d1212da9e0be82a12dbb1a8558e85","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-15 19:43:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9c45aa145ea69a2e00b2d7f072b1ec41b22c11b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:41:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/865a651414a6497ab8156e06865565e4 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93d06b3efc264b8fb475396e6ec140d9 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb4b462fd2ed42dfa7f4db53e29f51ef : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2037ed4df9241ca99e134f05da6fcdf : SUCCESS in 11m 16s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/04f8664ae9334d5cabd0126bd9c86644 : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d52a8fb0b6c4784b974d4368034dd6b : SUCCESS in 8m 05s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c38cfcb5c104d12a349d72a58454554 : SUCCESS in 7m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5a7e7b196ac471bbb913de817760374 : SUCCESS in 8m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/496ab2bba6a148d5a1e3ed193582b10f : FAILURE in 40m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cf12c112d6a4bd2941f60069e352e8d : SUCCESS in 1h 52m 14s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33a20e0be6964c0ea42f8e9e7390fb29 : SUCCESS in 1h 48m 35s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9b434fb5cea45b9afe5c63c064d0042 : SUCCESS in 26m 49s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/086659bf86bb440f8f31e50e009822d5 : FAILURE in 17m 34s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d5c2f47ec2f44a0a88a389014ad5a4f : SUCCESS in 24m 24s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bf7b2d132144e81b9b8578cd6957561 : FAILURE in 17m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb85b26b3b7143bb853eb6af021a2969 : FAILURE in 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad7314cfabb784583140037bddc3af337bc99077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:42:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d1938567efc885d7270095e6f78f4f45d8277a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 23:07:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd013093afce4054b44e401d2372a03e : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd71d9d06d9446da7e9373460e6ce6b : SUCCESS in 3m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1720653fb7314f9c817c10f82a2faa35 : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c7ff919e7104176bcf87357b18c5a49 : SUCCESS in 10m 26s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f79a7b44cbc4e6a8350e7a9bbe5dab7 : SUCCESS in 10m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cac3c780f7504b33b826790a7aa93421 : SUCCESS in 9m 24s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/399df5439b3d4d5c86dff67dead7e3e7 : SUCCESS in 9m 28s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/af2a5b224b6d408baa131c9a79a09d71 : SUCCESS in 1h 23m 20s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6328835f94b14f0d8f187b6668d966ab : SUCCESS in 23m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42681acf34c14d5faa2da0094aa746cc : FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7a24d43102398fe3a032fd340f304778a2b1e0b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 23:07:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cfbe2484cd5ce4b2b0933a18e78bfa1b46d6679","revisions":{"97f9a0bbce12c4640e597d018ed3bdfa742e3cc1":{"kind":"REWORK","_number":1,"created":"2021-06-14 07:45:03.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/39/796239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/39/796239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/1"}}},"commit":{"parents":[{"commit":"3e4d8d14c53b310e0f59d61b0e60054ecc971c30","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/3e4d8d14c53b310e0f59d61b0e60054ecc971c30"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-14 07:44:42.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-14 07:45:00.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I35b86a599d9e947648cad38904d54a8f998696e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/97f9a0bbce12c4640e597d018ed3bdfa742e3cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/97f9a0bbce12c4640e597d018ed3bdfa742e3cc1"}]},"branch":"refs/heads/master"},"0cfbe2484cd5ce4b2b0933a18e78bfa1b46d6679":{"kind":"REWORK","_number":2,"created":"2021-06-15 10:59:45.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/39/796239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/39/796239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/39/796239/2"}}},"commit":{"parents":[{"commit":"3e4d8d14c53b310e0f59d61b0e60054ecc971c30","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/3e4d8d14c53b310e0f59d61b0e60054ecc971c30"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-14 07:44:42.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-06-15 10:59:35.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I35b86a599d9e947648cad38904d54a8f998696e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/0cfbe2484cd5ce4b2b0933a18e78bfa1b46d6679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/0cfbe2484cd5ce4b2b0933a18e78bfa1b46d6679"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
