)]}'
{"id":"openstack%2Fpuppet-zaqar~829322","triplet_id":"openstack%2Fpuppet-zaqar~master~Ifb3f7866b477aadbbdf8470a059c2a2151db1135","project":"openstack/puppet-zaqar","branch":"master","topic":"c9-ut","hashtags":[],"change_id":"Ifb3f7866b477aadbbdf8470a059c2a2151db1135","subject":"DNM: Test unit tests for CentOS 9","status":"ABANDONED","created":"2022-02-15 15:32:18.000000000","updated":"2022-02-15 23:01:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94cb95ebbfcaf8e2d769295f2da733759943b63b","_number":829322,"virtual_id_number":829322,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-15 17:23:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 17:23:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ff86f26ecc72a1ee699108ff01f73ecd62debf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:32:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f3792f696c2feb96d7eeca7f3f928687ea013f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 16:40:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1b29e01be3547d2c2c43222e2217916b1ca6620c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 17:23:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9906f8b939234fa6a1232d278f541686 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43e0a319374042ce86f169957505c818 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5d1f52b0ae24e51a578aeae5f38073e : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/68ea53bf539d4415b3512af65c9b4281 : SUCCESS in 35m 11s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9e57cc62b2e4826b0c4ff92d063cf99 : SUCCESS in 29m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/550850c6b0004b20888e775b0ab8e95e : SUCCESS in 11m 10s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf3fe9200a74cad931e35527b6d4865 : SUCCESS in 12m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca1221ec88474695be00959f649bba0b : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94cb95ebbfcaf8e2d769295f2da733759943b63b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 23:01:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af0b2a1ef54ca63c6c0eb42de73d81da617c9ed9","revisions":{"f9782191d912a51f739b59d0a791b3bb53d26786":{"kind":"REWORK","_number":1,"created":"2022-02-15 15:32:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/829322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/22/829322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/1"}}},"commit":{"parents":[{"commit":"1e36ec00e51c66c93a9e93983189a964a0c95c30","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/1e36ec00e51c66c93a9e93983189a964a0c95c30"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:32:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:32:06.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\nChange-Id: Ifb3f7866b477aadbbdf8470a059c2a2151db1135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/f9782191d912a51f739b59d0a791b3bb53d26786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/f9782191d912a51f739b59d0a791b3bb53d26786"}]},"branch":"refs/heads/master"},"af0b2a1ef54ca63c6c0eb42de73d81da617c9ed9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-15 16:40:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/829322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/22/829322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/22/829322/2"}}},"commit":{"parents":[{"commit":"8ad1daf24f603f025bbf124480abaf9433a8eeef","subject":"CentOS 9: Fix unit tests for zaqar::wsgi::apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/8ad1daf24f603f025bbf124480abaf9433a8eeef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:32:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 16:40:04.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\nChange-Id: Ifb3f7866b477aadbbdf8470a059c2a2151db1135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/af0b2a1ef54ca63c6c0eb42de73d81da617c9ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/af0b2a1ef54ca63c6c0eb42de73d81da617c9ed9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
