)]}'
{"id":"openstack%2Fpuppet-zaqar~962073","triplet_id":"openstack%2Fpuppet-zaqar~master~Id545069b6766bbd367f119071234d80eb5fee9a5","project":"openstack/puppet-zaqar","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-25 17:01:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id545069b6766bbd367f119071234d80eb5fee9a5","subject":"Validate ensure parameter for package resources","status":"MERGED","created":"2025-09-23 08:58:13.000000000","updated":"2025-09-25 17:01:02.000000000","submitted":"2025-09-25 17:01:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962073","meta_rev_id":"e15b4060e671e15f579473ff15c7bcccd0a8e892","_number":962073,"virtual_id_number":962073,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 17:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-09-23 12:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-09-25 16:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-09-25 16:17:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-23 10:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 12:22:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-09-25 16:17:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d9bfc1e402f76a9c4d428ab6ecfef5e19bf9aa08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-23 08:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cb4a710358ba90c45ce21ce5d03e57681bb9cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 10:19:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e91b6dd55b934e3bba8774e04a9cadfc\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06d6d940112a4ed6a6b6b0639e8db315 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae504e16254e40fcafe1b567ebbe1824 : SUCCESS in 6m 02s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/bdf1b313ff29498dac9d18d7933241dd : SUCCESS in 3m 57s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f60fd31df6b941beb9d28dc0e8e6c47e : SUCCESS in 5m 35s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60917d7da47c4fcf8ed3132717c5f2c4 : SUCCESS in 4m 11s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/504eddbb9f24408b8b55685bed32717f : SUCCESS in 1h 04m 00s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cd42537a0ed48369ac5e2ac2ba29c3b : SUCCESS in 1h 16m 32s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74576b71453140c58fa18222e8e286aa : SUCCESS in 46m 35s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fd5f3fca2544886875ac1168ed1a603 : SUCCESS in 10m 35s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/492a12cc339c489794aa310458662693 : SUCCESS in 7m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/773720ed6fc64207991deaccdc16f51d : SUCCESS in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16f3614dec2893cd72170a358a8369c9a592d2e4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-23 12:22:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0fb1b97e501d14156e3f720fd4d80b8cfa9e3dbd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-09-25 16:17:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae99cc1ba8b9085325bc170f6dbd9271ba15d806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 16:18:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48963ea5140844804c189c11ea96d0af171dda66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 17:01:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c47247298d7445a193f295fda6655151\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98ca5e62e59f492ea3a27554e492f24b : SUCCESS in 5m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60ec0b67609743be8fab78c952effc0d : SUCCESS in 2m 50s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/b672b6726beb4a1d9aeae167bf9cc9db : SUCCESS in 3m 46s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79ead18c311d4d60839234addd96f635 : SUCCESS in 3m 07s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f8483dcf00e49ae9163ca5d03169a71 : SUCCESS in 41m 46s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d72855e66514828827aad7f6fdb3181 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e15b4060e671e15f579473ff15c7bcccd0a8e892","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 17:01:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"253c76869bf436eab4744d109bac574c8986f8b4","revisions":{"253c76869bf436eab4744d109bac574c8986f8b4":{"kind":"REWORK","_number":1,"created":"2025-09-23 08:58:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/962073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/73/962073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/73/962073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/73/962073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/73/962073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/73/962073/1"}}},"commit":{"parents":[{"commit":"549d2db4300244817bc8b509bb8b58e081a55bb8","subject":"Use new openstackclient tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/549d2db4300244817bc8b509bb8b58e081a55bb8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:57:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:57:59.000000000","tz":540},"subject":"Validate ensure parameter for package resources","message":"Validate ensure parameter for package resources\n\nThe minimum version of puppetlabs-stdlib has been bumped globally, so\nnow we can use the common type definition.\n\nChange-Id: Id545069b6766bbd367f119071234d80eb5fee9a5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/253c76869bf436eab4744d109bac574c8986f8b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/253c76869bf436eab4744d109bac574c8986f8b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
