)]}'
{"id":"openstack%2Fpuppet-zaqar~993263","triplet_id":"openstack%2Fpuppet-zaqar~master~Ib9e150eaee20e597c9f7de35d6b4a5ba0580fa49","project":"openstack/puppet-zaqar","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-30 06:13:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib9e150eaee20e597c9f7de35d6b4a5ba0580fa49","subject":"Add support for zaqar-gc cron job","status":"NEW","created":"2026-06-14 14:16:21.000000000","updated":"2026-06-30 06:13:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a7f3f213bff92d2ae30169fda11b273f85d0bc9","_number":993263,"virtual_id_number":993263,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-15 03:20:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2026-06-30 06:13:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-14 17:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 06:13:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9d3d67d3c0d471e4703c30eafd38e3d0b95e8a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-14 14:16:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb051c5cbfb4291520b77fa7e125613c1ee681fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-14 14:17:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16d9c7ccffd28b59c38cfdd8834bf6d3cba0b668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-14 15:31:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5dd1612f08935c3a24203c345ebc42bd3fec0445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 17:05:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e2386ec725a41efb8a3b370ad398ac1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38b1894d396a41e182803bd25d2676f4 : FAILURE in 2m 53s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9130ebc801604375b378499cc7ccd04b : SUCCESS in 3m 11s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/550c3af6d6234f819163e1d1d426a545 : FAILURE in 4m 03s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8809fe5885540e8a1350f842e009e6a : FAILURE in 5m 44s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6785456a52343cbb0de71a604292840 : FAILURE in 5m 45s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89d3353d505d429d9a8df41486d98a96 : FAILURE in 1h 15m 02s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e03dd02d174846b9a423996409e03e45 : SUCCESS in 1h 28m 09s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6336de7220d241459998a9187f81b514 : SUCCESS in 1h 13m 23s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6af448d9bd234516a6cc4a4abd03cab1 : SUCCESS in 14m 26s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5de9fc56bd6749299e7b9c3ac7fceeb0 : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7ab9cffd2c74567b4d9bce689aa638c : SUCCESS in 14m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85e215a7ed5844ceaff5a87b09c7da62 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3},{"id":"42495f41f03927b05bb5321affaaac90ba121629","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-15 01:53:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9eaffa8c8efac3cf62c57975e05a44776f3b3774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 03:20:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ac7d1ca2ae94148bac46ced8d37614d\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d99f771b4cfd4e4991c16dee06c86df7 : SUCCESS in 1m 42s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f8a9650d8374dbca59dc15c299c0ffb : SUCCESS in 2m 42s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/fc174c7620d6415fbc7e37e8c573f379 : SUCCESS in 4m 44s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0425f0daf07c4f56ab33fa113743d458 : SUCCESS in 4m 51s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2da7f80bab7146c584f524b151e2282e : SUCCESS in 6m 13s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fe594f9e72948f9a39166a46791a35b : FAILURE in 1h 11m 38s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/295d24e445c64c0ebc1dacb429d77a1b : SUCCESS in 1h 21m 25s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/951711dce4eb43babeef3ce707193ab0 : SUCCESS in 1h 19m 45s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9492c0a6898e4dc984b87aae144b79d0 : SUCCESS in 10m 29s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74be5b60f74a4d11a640d2795d2297de : SUCCESS in 16m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a047e8e17b7454cb72dcd1e5f9ee16c : SUCCESS in 11m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c6d58dedf9f4c418b63ac4fe06a6bd0 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":4},{"id":"8a7f3f213bff92d2ae30169fda11b273f85d0bc9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 06:13:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ffc9f9b1b16947aa0dd4af6d4ef4c680fccb6b43","revisions":{"e43d4bdb624208eb31d78b8c0ca5382e2f9f408a":{"kind":"REWORK","_number":1,"created":"2026-06-14 14:16:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/993263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/63/993263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/1"}}},"commit":{"parents":[{"commit":"c12f2f131f7b153babaed6fd3674c0d9f6968c71","subject":"Merge \"Bump upper version of puppetlabs-stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/c12f2f131f7b153babaed6fd3674c0d9f6968c71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 14:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 14:15:52.000000000","tz":540},"subject":"Add support for zaqar-gc cron job","message":"Add support for zaqar-gc cron job\n\nThe zaqar-gc tool should be run periodically to purge expired messages.\n\nChange-Id: Ib9e150eaee20e597c9f7de35d6b4a5ba0580fa49\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/e43d4bdb624208eb31d78b8c0ca5382e2f9f408a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/e43d4bdb624208eb31d78b8c0ca5382e2f9f408a"}]},"branch":"refs/heads/master"},"0f61cebca37e5d8e717b800b9d93f0c5e7b3b08b":{"kind":"REWORK","_number":2,"created":"2026-06-14 14:17:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/993263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/63/993263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/2"}}},"commit":{"parents":[{"commit":"c12f2f131f7b153babaed6fd3674c0d9f6968c71","subject":"Merge \"Bump upper version of puppetlabs-stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/c12f2f131f7b153babaed6fd3674c0d9f6968c71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 14:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 14:17:14.000000000","tz":540},"subject":"Add support for zaqar-gc cron job","message":"Add support for zaqar-gc cron job\n\nThe zaqar-gc tool should be run periodically to purge expired messages.\n\nChange-Id: Ib9e150eaee20e597c9f7de35d6b4a5ba0580fa49\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/0f61cebca37e5d8e717b800b9d93f0c5e7b3b08b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/0f61cebca37e5d8e717b800b9d93f0c5e7b3b08b"}]},"branch":"refs/heads/master"},"26c676ab8f15d1576ae96630487d8a19e30aebfe":{"kind":"REWORK","_number":3,"created":"2026-06-14 15:31:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/993263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/63/993263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/3"}}},"commit":{"parents":[{"commit":"c12f2f131f7b153babaed6fd3674c0d9f6968c71","subject":"Merge \"Bump upper version of puppetlabs-stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/c12f2f131f7b153babaed6fd3674c0d9f6968c71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 14:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:30:53.000000000","tz":540},"subject":"Add support for zaqar-gc cron job","message":"Add support for zaqar-gc cron job\n\nThe zaqar-gc tool should be run periodically to purge expired messages.\n\nChange-Id: Ib9e150eaee20e597c9f7de35d6b4a5ba0580fa49\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/26c676ab8f15d1576ae96630487d8a19e30aebfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/26c676ab8f15d1576ae96630487d8a19e30aebfe"}]},"branch":"refs/heads/master"},"ffc9f9b1b16947aa0dd4af6d4ef4c680fccb6b43":{"kind":"REWORK","_number":4,"created":"2026-06-15 01:53:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/993263/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-zaqar","ref":"refs/changes/63/993263/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-zaqar refs/changes/63/993263/4"}}},"commit":{"parents":[{"commit":"c12f2f131f7b153babaed6fd3674c0d9f6968c71","subject":"Merge \"Bump upper version of puppetlabs-stdlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/c12f2f131f7b153babaed6fd3674c0d9f6968c71"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 14:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-15 01:52:50.000000000","tz":540},"subject":"Add support for zaqar-gc cron job","message":"Add support for zaqar-gc cron job\n\nThe zaqar-gc tool should be run periodically to purge expired messages.\n\nChange-Id: Ib9e150eaee20e597c9f7de35d6b4a5ba0580fa49\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/ffc9f9b1b16947aa0dd4af6d4ef4c680fccb6b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-zaqar/commit/ffc9f9b1b16947aa0dd4af6d4ef4c680fccb6b43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
