)]}'
{"id":"openstack%2Fpycadf~667757","triplet_id":"openstack%2Fpycadf~master~I3555792aff294a1ea533464ef78a1cc8bf545c82","project":"openstack/pycadf","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I3555792aff294a1ea533464ef78a1cc8bf545c82","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-06-26 23:30:58.000000000","updated":"2019-07-08 21:31:20.000000000","submitted":"2019-07-08 21:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667757-1562621480991-13c00058","meta_rev_id":"05d67769be096104c38ae06ebdadc6d4eb7e5aaa","_number":667757,"virtual_id_number":667757,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2019-07-08 21:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-07-06 00:51:24.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-07-08 19:40:30.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"date":"2019-07-05 23:42:08.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2019-07-08 19:40:30.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 23:42:08.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-06 00:51:24.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-07-08 19:40:30.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-07-08 21:31:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdf75d51a242c7da0908dc1c9eb8d20dcf4c5ba4","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-26 23:30:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df02c93e12aba31d39b7c02a66c3655e06d92a2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 01:39:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/57/667757/1/check/tempest-full/592c7ae/ : SUCCESS in 1h 57m 51s\n- tempest-full-py3 http://logs.openstack.org/57/667757/1/check/tempest-full-py3/3771e5c/ : SUCCESS in 1h 41m 21s\n- openstack-tox-cover http://logs.openstack.org/57/667757/1/check/openstack-tox-cover/643ce67/cover/ : SUCCESS in 2m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/667757/1/check/openstack-tox-lower-constraints/6990ff8/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/57/667757/1/check/openstack-tox-pep8/ea19026/ : SUCCESS in 2m 34s\n- openstack-tox-py27 http://logs.openstack.org/57/667757/1/check/openstack-tox-py27/0fa444f/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/57/667757/1/check/openstack-tox-py36/6f150dd/ : SUCCESS in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/57/667757/1/check/openstack-tox-py37/b8bc190/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/57/667757/1/check/openstack-tox-docs/00e9aa2/ : POST_FAILURE in 2m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3518572688933c6e78c07fe7ca71ae696ad823d2","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-05 23:42:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"843979a3a8923296f6dc1969f2ce2f2f4142646f","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-07-06 00:51:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c5a885e47568297b525d6f85264e3c80a3c6d671","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 01:16:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/57/667757/2/check/tempest-full/1908827/ : SUCCESS in 1h 26m 17s\n- tempest-full-py3 http://logs.openstack.org/57/667757/2/check/tempest-full-py3/1f0289c/ : SUCCESS in 1h 32m 31s\n- openstack-tox-cover http://logs.openstack.org/57/667757/2/check/openstack-tox-cover/5905b1c/cover/ : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/667757/2/check/openstack-tox-lower-constraints/0c02134/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/57/667757/2/check/openstack-tox-pep8/20bb4e2/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/57/667757/2/check/openstack-tox-py27/9e33113/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/57/667757/2/check/openstack-tox-py36/71f5a55/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/57/667757/2/check/openstack-tox-py37/45dc4c7/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/57/667757/2/check/openstack-tox-docs/520aa2c/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3cf01d01822c86dc1b7ae15225c39a03d362e957","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-07-08 19:40:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"266aecaf331fd7850f488e1beafd06aa50af0c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 19:40:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e0d2966d0cef2eda4864a9a595d1b061c19b5cf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 21:31:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/57/667757/2/gate/tempest-full/d2577c5/ : SUCCESS in 1h 48m 10s\n- tempest-full-py3 http://logs.openstack.org/57/667757/2/gate/tempest-full-py3/841028d/ : SUCCESS in 1h 30m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/667757/2/gate/openstack-tox-lower-constraints/9b2e1ce/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/57/667757/2/gate/openstack-tox-pep8/d4fa674/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/57/667757/2/gate/openstack-tox-py27/f29378f/ : SUCCESS in 2m 47s\n- openstack-tox-py36 http://logs.openstack.org/57/667757/2/gate/openstack-tox-py36/1015588/ : SUCCESS in 3m 02s\n- openstack-tox-py37 http://logs.openstack.org/57/667757/2/gate/openstack-tox-py37/5474901/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/57/667757/2/gate/openstack-tox-docs/6e7c7ee/html/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8d3a0217edc6519cc8c352628c565f10a2768823","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 21:31:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c9e79e93f3c1e6d51644dd7b94eab18045e4daa","revisions":{"a52aa5e205ac4c92679516d0b64bd3e4c366910f":{"kind":"REWORK","_number":1,"created":"2019-06-26 23:30:58.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/57/667757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/57/667757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/57/667757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/57/667757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/57/667757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/57/667757/1"}}},"commit":{"parents":[{"commit":"8980f25545cb5e781dbe484b577f47455b775611","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/8980f25545cb5e781dbe484b577f47455b775611"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-26 22:20:36.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-26 23:30:57.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I3555792aff294a1ea533464ef78a1cc8bf545c82\nStory: #2005924\nTask: #34215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/a52aa5e205ac4c92679516d0b64bd3e4c366910f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/a52aa5e205ac4c92679516d0b64bd3e4c366910f"}]},"branch":"refs/heads/master"},"5c9e79e93f3c1e6d51644dd7b94eab18045e4daa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-05 23:42:08.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/57/667757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/57/667757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/57/667757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/57/667757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/57/667757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/57/667757/2"}}},"commit":{"parents":[{"commit":"8980f25545cb5e781dbe484b577f47455b775611","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/8980f25545cb5e781dbe484b577f47455b775611"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-26 22:20:36.000000000","tz":-240},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-05 23:41:57.000000000","tz":-420},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nDepends-On: https://review.opendev.org/669543\n\nChange-Id: I3555792aff294a1ea533464ef78a1cc8bf545c82\nStory: #2005924\nTask: #34215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/5c9e79e93f3c1e6d51644dd7b94eab18045e4daa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/5c9e79e93f3c1e6d51644dd7b94eab18045e4daa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
