)]}'
{"id":"openstack%2Fpycadf~685432","triplet_id":"openstack%2Fpycadf~master~I83c2aa99da035c6c7979eda19ca655f48fc4122a","project":"openstack/pycadf","branch":"master","topic":"685432","hashtags":[],"change_id":"I83c2aa99da035c6c7979eda19ca655f48fc4122a","subject":"Follow the PTI for docs","status":"ABANDONED","created":"2019-09-27 17:15:30.000000000","updated":"2022-01-21 20:56:38.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"799cc6a612879453516875c2c731721070710b86","_number":685432,"virtual_id_number":685432,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"date":"2019-10-14 07:42:41.000000000","_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"date":"2019-10-09 09:27:02.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-09 11:18:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":-1,"date":"2019-10-11 21:17:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 09:27:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-09 11:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-11 21:17:35.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"}],"messages":[{"id":"bffac895007d845d16682b7789608340cb0a8410","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-27 17:15:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"205b95d32c0e22902c88fb10d36e552c2b4cbba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:20:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea0450b579014e018581a85494ab6972 : FAILURE in 2m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7177bcf3e3c547a8b9a32f9137f2ff13 : SUCCESS in 1h 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2e2838264b34115a7a0796ea4d6d60e : SUCCESS in 1h 29m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72663ec2002b4549a5570ad36ce657d7 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a59fbd1516104fd4aa5a33bf5f24d1b3 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60bc6c0cd914e4aaa21e680a515a9a7 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dde0938b60de44f38b1fd3e080f9ce8c : SUCCESS in 2m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dea06ecd6bae43da9b1672c589e29055 : SUCCESS in 2m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a8fb5f4c94f4545aaeaa86ae8d329b6 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75220597b034b3bbd0bbd82757b5cba : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8ca6c8c70ac0366079af05ba23c661b6e0895be8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-30 11:19:04.000000000","message":"Patch Set 1: Code-Review-1\n\nYou need to update lower-constraints too","accounts_in_message":[],"_revision_number":1},{"id":"075010fc4b5ba1c02948569c66efd42c2113a729","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-09 09:15:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f25b9570aafc1ee57d9b093cf8a027b613a49e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-09 09:27:02.000000000","message":"Patch Set 2:\n\nThink you still need to update lower-constraints, no?","accounts_in_message":[],"_revision_number":2},{"id":"523df549f5c994d31ed6cb705276659568f9361a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-09 09:34:25.000000000","message":"Patch Set 2:\n\n\u003e Think you still need to update lower-constraints, no?\n\nRemoved sphinxcontrib.apidoc because we\u0027re not configuring autodoc in setup.cfg. Leaving this out and doing the minimum needed to get sphinx-build to work.","accounts_in_message":[],"_revision_number":2},{"id":"f6e469438b0564dd8b440a4b9b6fcd2ac85b47ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:18:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2508d9015a594431ac0e0d719b989b16 : SUCCESS in 2m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e716b8ee309b46c986f5891c80891023 : SUCCESS in 1h 56m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41614291f69140e981ed8f3a798820ed : SUCCESS in 1h 43m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/260b776b16ca476ab7ff2ef1f9a17dd2 : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06c1099d451341e7881692ce9fc557af : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ffa9983fc84bc2be87de2022a5a429 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f349fd3b612649a79039e51b56213dcb : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b53d6118d2dd439dbee12a12fb978275 : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40a300d2c2194fdaa4c3b65638e9ad17 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e9ec80b6d147eab5bada66c3cc8d21 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"26b04cf305c1295e147cf6a5fc5ebad0706d73f2","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-11 21:17:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"770124112f5641e3e2667ec1b7d85d1223def063","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-14 07:42:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53afd778cd1de028f2bd79eefe2a9526e23924bc","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-14 18:53:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"799cc6a612879453516875c2c731721070710b86","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 20:56:38.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5d4c3844539766e2e58b3a3659ae6152e5f56af","revisions":{"f5bbf3efdda2903aae3effa186e24993ded9c6f6":{"kind":"REWORK","_number":1,"created":"2019-09-27 17:15:30.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/32/685432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/32/685432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/32/685432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/32/685432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/32/685432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/32/685432/1"}}},"commit":{"parents":[{"commit":"d113c1564d49d93451d9330955e8a42b7db04149","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/d113c1564d49d93451d9330955e8a42b7db04149"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-27 17:15:02.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-27 17:15:02.000000000","tz":330},"subject":"Follow the PTI for docs","message":"Follow the PTI for docs\n\nUse sphinx-build instead of the pbr sphinx extention for building docs\nas instructed by the PTI[1].\n\nThis requires using the sphinxcontrib-apidoc plugin rather than the\nautodoc pbr extention. We also remove the reference to the ChangeLog\nfile that is usually generated by pbr and instead refer to the published\nreno release notes. Also fixes the header formatting for the index page,\nas the headers weren\u0027t rendering at all.\n\n[1] https://governance.openstack.org/tc/reference/pti/python.html\n\nChange-Id: I83c2aa99da035c6c7979eda19ca655f48fc4122a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/f5bbf3efdda2903aae3effa186e24993ded9c6f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/f5bbf3efdda2903aae3effa186e24993ded9c6f6"}]},"branch":"refs/heads/master"},"b5d4c3844539766e2e58b3a3659ae6152e5f56af":{"kind":"REWORK","_number":2,"created":"2019-10-09 09:15:38.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/32/685432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/32/685432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/32/685432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/32/685432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/32/685432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/32/685432/2"}}},"commit":{"parents":[{"commit":"d113c1564d49d93451d9330955e8a42b7db04149","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/d113c1564d49d93451d9330955e8a42b7db04149"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-27 17:15:02.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-10-09 09:14:37.000000000","tz":330},"subject":"Follow the PTI for docs","message":"Follow the PTI for docs\n\nUse sphinx-build instead of the pbr sphinx extention for building docs\nas instructed by the PTI[1].\n\nWe also remove the reference to the ChangeLog file that is usually\ngenerated by pbr and instead refer to the published reno release notes.\nAlso fixes the header formatting for the index page, as the headers\nweren\u0027t rendering at all.\n\n[1] https://governance.openstack.org/tc/reference/pti/python.html\n\nChange-Id: I83c2aa99da035c6c7979eda19ca655f48fc4122a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/b5d4c3844539766e2e58b3a3659ae6152e5f56af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/b5d4c3844539766e2e58b3a3659ae6152e5f56af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
