)]}'
{"id":"openstack%2Fpycadf~691159","triplet_id":"openstack%2Fpycadf~master~Ice4c614cd2bf239098753ed76ecec41c9bb60b42","project":"openstack/pycadf","branch":"master","hashtags":[],"change_id":"Ice4c614cd2bf239098753ed76ecec41c9bb60b42","subject":"Switch to Ussuri jobs","status":"MERGED","created":"2019-10-25 02:51:24.000000000","updated":"2019-12-19 00:13:45.000000000","submitted":"2019-12-19 00:10:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691159-1576714223878-71032318","meta_rev_id":"0e01a11f6aa66b337fdf9031a866a86818edae58","_number":691159,"virtual_id_number":691159,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 00:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-12-18 22:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-10-25 04:13:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-12-18 22:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 04:13:18.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-12-18 22:32:13.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-19 00:10:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6558e0a6ef8c9b4018ef2a22ca1c86a13ab350b7","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-25 02:51:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"581cb610854075e67cd148d5a35040fcaf4d43dd","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-25 04:13:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"93f1cbac25e9d2bd4c3793aec452110e0f2cc76d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 04:41:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ab952403460441a9aa60427cb119369a : SUCCESS in 1h 44m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b07eb71f1d64fd093360e2a822f549a : SUCCESS in 1h 38m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e2d24fdb76f437490011feef579ecc1 : SUCCESS in 2m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/791a47fd984541a5bbb01bde0de23905 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ad6401bb534364b96fa841a0dadac5 : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58b77ea74c8440a4805669d9338424db : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d47e001714d457c8dff39e9c7e41429 : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8b0cdd9e0204150810fac55343442db : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38a9f3e8ef8449d98b6e3cf9475b717 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cc223d9524c66e54ebcb3f3d50f5c214e948995d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-18 22:32:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSingle-approving as this is a mass-proposed trivial patch that has been open for two months.","accounts_in_message":[],"_revision_number":1},{"id":"ed0baab3ab1a548d478dc190d22ce1a8227e9049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 22:32:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97daf8e9202d7f20d02cfeeb02a79a00ecba6de2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 00:10:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0aa4fd455e4e4925a4219dd63fe810b4 : SUCCESS in 1h 31m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc47254cf2794deaac06d8345a2d0e38 : SUCCESS in 1h 31m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ce3d9b48c6149c696c14e2a0aa5b011 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9735920877d44d4b857ea04b55f01d9f : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03377ade0062422c9368dacc9df69693 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c08d1f8d3544af790211a1a8a9c1f4d : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ca8c39ce8ff43dbbc3a08edc42f674e : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb61e4f4f814e01aea42d6f86d1985d : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e59282d698d9dce56e30db48f263f5406a5be919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 00:10:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0e01a11f6aa66b337fdf9031a866a86818edae58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 00:13:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca396abcfac34c1ba419dfd5173d2b3d : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"885a7093a02ea0b5d466befbcd37821efdb861bf","revisions":{"885a7093a02ea0b5d466befbcd37821efdb861bf":{"kind":"REWORK","_number":1,"created":"2019-10-25 02:51:24.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/59/691159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/59/691159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/59/691159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/59/691159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/59/691159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/59/691159/1"}}},"commit":{"parents":[{"commit":"fc7d8debee781f457aca1860196e86a81a8eb329","subject":"Update source link in readme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/fc7d8debee781f457aca1860196e86a81a8eb329"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-25 02:51:19.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-25 02:51:19.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: Ice4c614cd2bf239098753ed76ecec41c9bb60b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/885a7093a02ea0b5d466befbcd37821efdb861bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/885a7093a02ea0b5d466befbcd37821efdb861bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
