)]}'
{"id":"openstack%2Fpycadf~750341","triplet_id":"openstack%2Fpycadf~master~I43b912be6900d3d8f081ae4065789278c97db0b0","project":"openstack/pycadf","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"I43b912be6900d3d8f081ae4065789278c97db0b0","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-08 12:28:09.000000000","updated":"2020-11-05 22:09:30.000000000","submitted":"2020-11-05 22:05:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750341-1604613940234-cb4ec074","meta_rev_id":"5a9f15503ac4a99d31152d3581ab78a862d10853","_number":750341,"virtual_id_number":750341,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 22:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-11-05 20:35:24.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-09-14 12:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-11-05 20:35:24.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 12:06:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-05 20:35:24.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-11-05 22:05:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57ec60f305b2ce0cd629ce9b913df00733d7df64","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-08 12:28:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fc6ac080f4e3581bb5c69800f51a5519fcc35e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:59:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/198dc8d5bf5743ec9d7031146a71dfc6 : SUCCESS in 1h 22m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2916f826ef249cbb18a27632fea2142 : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1c80c6a71d34c51a37c510e0bdbc8f8 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a7fc8e39d4483cbdf78a5801f3e6cf : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b39fc9a38d4739a81ae3ba604b3a53 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf6c9a6b705242bcacb4a13a504d80ed : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e281b17135504b86bb32ddce75cdcb86 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7a69ecdfee4f1a0428a9d527c3b39e54cc7e86cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 12:06:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d4621ab0be0c0532d203658e721805936904734","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-11-05 20:35:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54d81ceb63b4acd8c0c6e850e46577ceed202d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 20:35:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"007f0dc1a8e8dc3e441a9f9da5b88ecb86bd7095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:05:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fd9601b575941698f9392f1a569502a : SUCCESS in 1h 22m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c92cae05c68145a2b407b77e0a1ad86b : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd85edfc28594b4bb72d9aaa7b09db59 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e24216b822046d795f9ff24f6e6e3d9 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d05172a58e5b454c8a6f75d2ed6e4f69 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b743f1d56c4abf959a9d3f5e37da7e : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4cb4f8305c641ef9cd568c05c984eded0c600e89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:05:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b4e879046454fb424d9da21e24a5f45e527b5f81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:09:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa016626c1044f1a0e1c8d6d0094ac4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b697886a094444f9cd631634307eea2de713751","revisions":{"2b697886a094444f9cd631634307eea2de713751":{"kind":"REWORK","_number":1,"created":"2020-09-08 12:28:09.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/41/750341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/41/750341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/41/750341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/41/750341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/41/750341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/41/750341/1"}}},"commit":{"parents":[{"commit":"4179996d52dbd37f3fd966225b4157343a75239c","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/4179996d52dbd37f3fd966225b4157343a75239c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-08 12:28:06.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-08 12:28:06.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I43b912be6900d3d8f081ae4065789278c97db0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/2b697886a094444f9cd631634307eea2de713751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/2b697886a094444f9cd631634307eea2de713751"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
