)]}'
{"id":"openstack%2Fpycadf~850842","triplet_id":"openstack%2Fpycadf~master~I149fdaae0db282930bd4245eace234072d8af705","project":"openstack/pycadf","branch":"master","attention_set":{},"removed_from_attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2022-08-17 09:17:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I149fdaae0db282930bd4245eace234072d8af705","subject":"Use TOX_CONSTRAINTS_FILE","status":"ABANDONED","created":"2022-07-25 09:01:59.000000000","updated":"2022-08-17 09:22:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13e0d804806e9ae58b1299f0126ee44166d0869e","_number":850842,"virtual_id_number":850842,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-25 10:27:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 10:27:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65f0c7841c988c67d98fb6c04ed70b127be523e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-07-25 09:01:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76f84194e892842ddd2efde27292e2bc0d3ab276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 10:27:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ef0f88006cf40b493667510d0163e18\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0dc3290117f4ffe8a6945eff9fa6d7e : SUCCESS in 1h 23m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b392fbe59164cbe8de86e1f10638eb2 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5c920943a24f369c6645fdcccf3a04 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79de8a199f7749e39b8372fe5941ae27 : FAILURE in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96560cd289e4493ba514aedce34029ce : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/875f8976cf774b508b6f77c92152640a : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4848296df734510b7d3bb4c69247247 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"867239a3d364c04ed849cc83c1f3f9b78c4ca1a8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-17 09:17:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8fec325b9c62bb465ba52b0c69203b7d4bcb7fda","tag":"autogenerated:gerrit:restore","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-17 09:22:37.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"13e0d804806e9ae58b1299f0126ee44166d0869e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-17 09:22:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6aec0143343711d91b99d4f773d7e4ee761eae62","revisions":{"6aec0143343711d91b99d4f773d7e4ee761eae62":{"kind":"REWORK","_number":1,"created":"2022-07-25 09:01:59.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/42/850842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/42/850842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/42/850842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/42/850842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/42/850842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/42/850842/1"}}},"commit":{"parents":[{"commit":"84f6e5b7658adc36e9e9cedcd42642c9ab8ceb15","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/84f6e5b7658adc36e9e9cedcd42642c9ab8ceb15"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-07-25 09:00:04.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-07-25 09:00:19.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I149fdaae0db282930bd4245eace234072d8af705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/6aec0143343711d91b99d4f773d7e4ee761eae62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/6aec0143343711d91b99d4f773d7e4ee761eae62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
