)]}'
{"id":"openstack%2Fpycadf~863691","triplet_id":"openstack%2Fpycadf~stable%2Fussuri~Ia3f66b6805704c254295fd663f0260d5d963ed66","project":"openstack/pycadf","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-11-28 10:15:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3f66b6805704c254295fd663f0260d5d963ed66","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-11-04 18:51:25.000000000","updated":"2022-11-28 10:18:34.000000000","submitted":"2022-11-28 10:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863691","meta_rev_id":"4a8b7647f86d6b717bc389dc9483cc90d8fe3d09","_number":863691,"virtual_id_number":863691,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-28 10:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-11-04 21:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-11-04 20:08:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2022-11-28 08:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 18:51:25.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-11-04 18:51:47.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-11-04 20:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 08:29:53.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"a36f2d16a93515f5cfacdc85eb3957f7c88b3a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 18:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a00e22c824bd03b00403e47264f4e4d9e798bf33","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 18:51:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ccd1c60433e2a29fef6e9824f53fdf19be9a2558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 19:39:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62ba967cd20f4c0be381807c6de7c6b91f8fd2f7","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 20:08:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"af2f202a22392e2b1c6635687222ed5de3a02984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 20:55:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0100f52eb0ca481b99b266af5b530f7a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2f6f71b7b424ba3a447e34c86b6aa16 : SUCCESS in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e32ae67d26a841cabc222fa9b63d8733 : SUCCESS in 3m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee4a2b195a6b40ad95d35dec28802d2c : SUCCESS in 1h 14m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47db349d3cef4194a254511fd90007e8 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde8976795194aa0b1307d436428e3bf : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fb6356ec395489bbe846a864c246a72 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01b4d7cc3b314d08aa03e91dcd3bdc0a : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c481bacb494719a2153a46b6530e4a : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ba1d9879c5fea149359f6205f241100803687d00","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 21:09:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"211e2053ec78c2ee9def01f352ecfd0446a91f00","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-11-28 08:29:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90f965b08c05d99371115bf3905c8e25a93d414a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 08:30:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"830b8790acc757741a32d9cc7e8b49cbfcc86e07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 10:15:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d199d8635e458db426b569e6c611f3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/74fcc8d5e8ad4e64b3567ca3155496de : SUCCESS in 2m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/230b7a6b142e45cfbdb07baa65e553f0 : SUCCESS in 1m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f44a1e0591144819c17be7b2c7983e9 : SUCCESS in 1h 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4730d40dd8ba438d9464fd43c536d4ed : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14625f5c7af244f3a6bb9149a8854cc9 : SUCCESS in 2m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2735ea90572945af9f51058143a3f805 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f316734ac81c437fa27c555924df7631 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a1c9074dc8e4d08174b740268e863dc64ded4abe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 10:15:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4a8b7647f86d6b717bc389dc9483cc90d8fe3d09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 10:18:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d6bfe086eda4ac7a5197c680ffbb749\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f49754b51b44eba7f92e572e0dd07b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6fe11b5188bbdbb39bff511fa0227e5785be17f7","revisions":{"414bc907b3d0440238dbf602d2b74e92364147c1":{"kind":"REWORK","_number":1,"created":"2022-11-04 18:51:25.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/91/863691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/91/863691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/91/863691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/91/863691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/91/863691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/91/863691/1"}}},"commit":{"parents":[{"commit":"66a84728aa566c4a0315cca00ebb9dc202edda88","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/66a84728aa566c4a0315cca00ebb9dc202edda88"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 22:13:35.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-11-04 18:51:11.000000000","tz":-240},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\n(cherry picked from commit 0475709c0734d8314637298de2fd2b005ad5a1cb)\n\nChange-Id: Ia3f66b6805704c254295fd663f0260d5d963ed66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/414bc907b3d0440238dbf602d2b74e92364147c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/414bc907b3d0440238dbf602d2b74e92364147c1"}]},"branch":"refs/heads/stable/ussuri"},"6fe11b5188bbdbb39bff511fa0227e5785be17f7":{"kind":"REWORK","_number":2,"created":"2022-11-04 19:39:52.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/91/863691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/91/863691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/91/863691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/91/863691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/91/863691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/91/863691/2"}}},"commit":{"parents":[{"commit":"66a84728aa566c4a0315cca00ebb9dc202edda88","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/66a84728aa566c4a0315cca00ebb9dc202edda88"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 22:13:35.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-11-04 19:39:47.000000000","tz":-240},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\n(cherry picked from commit 0475709c0734d8314637298de2fd2b005ad5a1cb)\n\nChange-Id: Ia3f66b6805704c254295fd663f0260d5d963ed66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/6fe11b5188bbdbb39bff511fa0227e5785be17f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/6fe11b5188bbdbb39bff511fa0227e5785be17f7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[]}
