)]}'
{"id":"openstack%2Fpycadf~863705","triplet_id":"openstack%2Fpycadf~stable%2Ftrain~I8bba952b111786fdda303aac2c517d723dd24330","project":"openstack/pycadf","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-11-04 22:03:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8bba952b111786fdda303aac2c517d723dd24330","subject":"Update setup.cfg classifiers for stable/train","status":"ABANDONED","created":"2022-11-04 19:53:07.000000000","updated":"2022-11-04 22:03:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"090e0948f9a6b2d340eba0b47d7970ca0d9e5cfd","_number":863705,"virtual_id_number":863705,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-04 21:13:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2022-11-04 20:07:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 20:07:48.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-11-04 21:13:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4de627a33489aa7bb70ac5d19861a854f6054133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 19:53:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60d9416ee8b53a4f19d0637a41c653ffc35a7026","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 20:07:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8e8c696674a3cb5c3210c93834ed960f5d9b0da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 21:13:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86bdb0b4cf09461bb3955d472dda6acc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f1075487bf8477da3f8425aed3cfb65 : SUCCESS in 1m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1dc9efd0e58a4328a6df4c76bf2a23ff : SUCCESS in 1h 05m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c79e1598c7b744d19cc81ebe464a9e13 : SUCCESS in 1h 19m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f0bff230b9349e5b52f4205151ab12c : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd1db1070bd74881850d82ec7411a248 : FAILURE in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af84e1f9a9b4b0e806fb5f9682380ef : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46761bd8656540f989e306da0895f690 : SUCCESS in 2m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/446e689bc5fa4cab8659bd037c9c53aa : SUCCESS in 2m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f4bbe8f3784458c9d3fed68d2102c93 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7601cce9f944c13a407a36b0691410c : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"090e0948f9a6b2d340eba0b47d7970ca0d9e5cfd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 22:03:24.000000000","message":"Abandoned\n\nThis change is now part of https://review.opendev.org/c/openstack/pycadf/+/863715","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed638315250299469fcf6fea44e63cc2bc41f97f","revisions":{"ed638315250299469fcf6fea44e63cc2bc41f97f":{"kind":"REWORK","_number":1,"created":"2022-11-04 19:53:07.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/05/863705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/05/863705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/05/863705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/05/863705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/05/863705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/05/863705/1"}}},"commit":{"parents":[{"commit":"22124cece7596c02d8a19a0f4c3088ec25941c61","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/22124cece7596c02d8a19a0f4c3088ec25941c61"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-11-04 19:53:03.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-11-04 19:53:03.000000000","tz":-240},"subject":"Update setup.cfg classifiers for stable/train","message":"Update setup.cfg classifiers for stable/train\n\nChange-Id: I8bba952b111786fdda303aac2c517d723dd24330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/ed638315250299469fcf6fea44e63cc2bc41f97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/ed638315250299469fcf6fea44e63cc2bc41f97f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
