)]}'
{"id":"openstack%2Fpycadf~863715","triplet_id":"openstack%2Fpycadf~stable%2Ftrain~Ia3f66b6805704c254295fd663f0260d5d963ed66","project":"openstack/pycadf","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-12-21 14:13:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia3f66b6805704c254295fd663f0260d5d963ed66","subject":"Drop lower-constraints.txt and its testing","status":"ABANDONED","created":"2022-11-04 22:02:43.000000000","updated":"2023-12-21 14:13:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d916a1ed9fef5f1dcf087fbc00b173584327a20","_number":863715,"virtual_id_number":863715,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-04 23:25:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 22:02:43.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-11-04 23:25:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b32451e09562ca5a895e04a67192acf3f4783cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 22:02:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beaab00f8ba1cf2dea8520d914e6948949bda825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 23:25:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9e98e937b0c46f494a00c1a7705e363\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0f6061966f104ac3988f746ea2add0e5 : SUCCESS in 1m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/64f9be2f17854b2ebb5498100d9626f4 : FAILURE in 1m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c87c413a4b0a4fc8a5b7a4a1c8bc5709 : SUCCESS in 1h 21m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76675cbd19874d8e8d463b030120ca66 : SUCCESS in 52m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfa9b2ff2cea407f9f325db4abb75185 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb8f006193441d7989ba68cea7f3a73 : SUCCESS in 2m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a11c8a470b704969b9dc82e3a632bb3e : SUCCESS in 2m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52785a9c7ecd43049e5202bfb078d977 : SUCCESS in 2m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad1efd9098ed45b680cde01525c872eb : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5503143a44495b967c6ac1109d269a : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6d916a1ed9fef5f1dcf087fbc00b173584327a20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:13:02.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55d02aa19e92336e17e038836668245079c6566d","revisions":{"55d02aa19e92336e17e038836668245079c6566d":{"kind":"REWORK","_number":1,"created":"2022-11-04 22:02:43.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/15/863715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/15/863715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/15/863715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/15/863715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/15/863715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/15/863715/1"}}},"commit":{"parents":[{"commit":"22124cece7596c02d8a19a0f4c3088ec25941c61","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/22124cece7596c02d8a19a0f4c3088ec25941c61"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 22:13:35.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-11-04 22:02:25.000000000","tz":-240},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\n(cherry picked from commit 6fe11b5188bbdbb39bff511fa0227e5785be17f7)\n\nChange-Id: Ia3f66b6805704c254295fd663f0260d5d963ed66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/55d02aa19e92336e17e038836668245079c6566d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/55d02aa19e92336e17e038836668245079c6566d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
