)]}'
{"id":"openstack%2Fpycadf~881248","triplet_id":"openstack%2Fpycadf~master~I44f290fe7246ffbf0689cd49df45554712dc274e","project":"openstack/pycadf","branch":"master","topic":"docs_no_rm","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-04-22 00:59:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I44f290fe7246ffbf0689cd49df45554712dc274e","subject":"Replate whitelist with allowlist","status":"ABANDONED","created":"2023-04-21 17:06:24.000000000","updated":"2023-04-22 00:59:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1fbd221f6d1254c5ff5442a641445997b15de0d","_number":881248,"virtual_id_number":881248,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-21 18:50:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-21 18:50:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2138fa94cc99b9a6b030eac047adbd27a9e8261b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-21 17:06:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7aa26822c27ce85e800fe22d21b0c396a0d145d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 18:50:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f2447fabdb74fcdb559f75991d7bba8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00f3894d3a604bd69ae560527bbb0512 : SUCCESS in 1h 29m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1612200553814c1da92a4bd2df1d0d00 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5217e33e536a4f1587cc2702225325b1 : FAILURE in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4dae1376bac4b6a89daa43c8315a288 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acb79ad17839487996af68925db7e213 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa15a71d3b84d4ca5903ac97e1e0eba : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e1fbd221f6d1254c5ff5442a641445997b15de0d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-22 00:59:22.000000000","message":"Abandoned\n\nFolded into Id12f6e3b5d603531744d02f7bafab0b80696751e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6af95656a64f7ebe58acf974be5122b2e58de650","revisions":{"6af95656a64f7ebe58acf974be5122b2e58de650":{"kind":"REWORK","_number":1,"created":"2023-04-21 17:06:24.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/48/881248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/48/881248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/48/881248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/48/881248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/48/881248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/48/881248/1"}}},"commit":{"parents":[{"commit":"bb4cc27a0ca9efbb0df04a08901fee088d47a536","subject":"Merge \"Add doc/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/bb4cc27a0ca9efbb0df04a08901fee088d47a536"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-04-21 17:01:26.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-04-21 17:06:16.000000000","tz":-300},"subject":"Replate whitelist with allowlist","message":"Replate whitelist with allowlist\n\nSomeone upgraded tox at the gate and our jobs now fail\nwith the following message:\n\nE failed with rm is not allowed, use allowlist_externals to allow it\n\nThe most practical choice is to remain enslaved by the\ngate master here. The patch simply renames the key.\n\nChange-Id: I44f290fe7246ffbf0689cd49df45554712dc274e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/6af95656a64f7ebe58acf974be5122b2e58de650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/6af95656a64f7ebe58acf974be5122b2e58de650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
