)]}'
{"id":"openstack%2Fpycadf~906649","triplet_id":"openstack%2Fpycadf~master~I268ae097eecd4c3562b1005658cf2927579659d0","project":"openstack/pycadf","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-29 16:42:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I268ae097eecd4c3562b1005658cf2927579659d0","subject":"Bump hacking","status":"MERGED","created":"2024-01-25 13:23:42.000000000","updated":"2024-01-29 16:44:42.000000000","submitted":"2024-01-29 16:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906649-bump-hacking","meta_rev_id":"abac74bddc87f92027df21f4676ff95dbb4f8ef9","_number":906649,"virtual_id_number":906649,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-29 16:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-01-29 09:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-01-29 09:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 14:51:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-29 09:39:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"afe1fd7e30eba9489e9bdda2efa618de7141f501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 13:23:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c256d68eb0e764f6eaf11026b82385d63f6b4b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 14:51:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8dc32ed3f834c2bb122de88d9e30a36\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0ea83739d0f342dd9ea8f0c3587df73a : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/928b3daf29a444fb9e52561ea76740a2 : FAILURE in 1h 19m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e5553e4a50e45249524af7649330b33 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b9b1b0198694150993013c8955ed3b3 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60896500e63e4d0f8d7bbe79fd7f41d7 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48e7bb372428446298832a418ded88b6 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b68992e58b44bcbee2b551e70174de : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9218bc24fe5ebc5c219269964cf2d61f0f883ab7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 14:55:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"466619d9650a403fbdbd785098a24721ece776d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 16:53:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e48f3f522b647c199b7f735959240eb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6770270e2704b729ef2ece4c46237eb : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9bb478a4a6a44cc9feb28e4ced19268 : SUCCESS in 1h 48m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c4d6634df1f4619af9b79b5ed022a63 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa3cfce57d27492d869a16cef75c23c2 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7295d4b8cb6417487ff57a1afd4fd19 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9d92679cbd4473586c23499c7312dad : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0530af098feb4b178fe50cc93653d181 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ca70c666a4f834f34b4829a0faa4979723a4cc34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-26 01:02:28.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"aa08be7cb537506903707f50b2f0cd8fae63ad06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 02:50:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b80718f0628459ca4d57b1b9c00150d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/780a9be65f6c4b7184456c36d3abb071 : SUCCESS in 2m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27c9279cb1184d98b3831975549a8ce4 : SUCCESS in 1h 39m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41fbf5cc48cc4460b3151995705ad583 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e31d4e2d7164a758f1f4aa58d67d166 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ce2f83d9ffa48c0acca10f901ec428e : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bee0e7c50af848b3b101c05d7a9728aa : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac143ce72ba46f1ab095cef1541d119 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"995d6fcdc03492332a3490aa6b0a01824270e670","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-29 09:39:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f2eefaf592b912f880bea9bbf2c55e31c695b3c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 15:02:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b80399b853e6937c453ccce1105567fa73ae3ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 16:42:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b40690dd5642f8923e192d8674fa5d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/abe80246534547ec9a578b826c7af2b0 : SUCCESS in 1m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b637c22fae3443419cc0078325fb9351 : SUCCESS in 1h 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5389ae4ccb504bb18a0cf5f30b059130 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2707fed6eaf34c8dbee728f1ec342721 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a9be74d382d45d6b41aaf933484646e : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de8acd9ac6a485ab7746787564d4507 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b7e58b8d349a01e2d6797ede356f414e5ee640db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 16:42:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"abac74bddc87f92027df21f4676ff95dbb4f8ef9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 16:44:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0585c2db911f4ff587fd71f4856715d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac54ed310e74ea38465f75560974d5b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ddd79e2afcd246908e71641ab60e929f5a7c5b69","revisions":{"44e9a65cb5d071e556d5c5a8dae9a91b617bf614":{"kind":"REWORK","_number":1,"created":"2024-01-25 13:23:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/906649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/49/906649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/49/906649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/49/906649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/49/906649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/49/906649/1"}}},"commit":{"parents":[{"commit":"34aa8194af13969f906fa868f7a916c4fa52a675","subject":"Remove usage of six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/34aa8194af13969f906fa868f7a916c4fa52a675"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 13:18:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 13:23:16.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\n... because hacking 3.0.x is too old.\n\nA few checks are disabled because these detect invalid problems.\n\nChange-Id: I268ae097eecd4c3562b1005658cf2927579659d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/44e9a65cb5d071e556d5c5a8dae9a91b617bf614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/44e9a65cb5d071e556d5c5a8dae9a91b617bf614"}]},"branch":"refs/heads/master"},"ddd79e2afcd246908e71641ab60e929f5a7c5b69":{"kind":"REWORK","_number":2,"created":"2024-01-26 01:02:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/906649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/49/906649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/49/906649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/49/906649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/49/906649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/49/906649/2"}}},"commit":{"parents":[{"commit":"34aa8194af13969f906fa868f7a916c4fa52a675","subject":"Remove usage of six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/34aa8194af13969f906fa868f7a916c4fa52a675"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 13:18:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-26 01:02:28.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\n... because hacking 3.0.x is too old.\n\nA few checks are disabled because these detect invalid problems.\n\nChange-Id: I268ae097eecd4c3562b1005658cf2927579659d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/ddd79e2afcd246908e71641ab60e929f5a7c5b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/ddd79e2afcd246908e71641ab60e929f5a7c5b69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
