)]}'
{"id":"openstack%2Fpycadf~930970","triplet_id":"openstack%2Fpycadf~master~I7df0e750c5e678e5d150382c23c969ca820650b9","project":"openstack/pycadf","branch":"master","hashtags":[],"change_id":"I7df0e750c5e678e5d150382c23c969ca820650b9","subject":"Remove unused debtcollector","status":"MERGED","created":"2024-10-01 02:50:29.000000000","updated":"2024-10-15 19:26:52.000000000","submitted":"2024-10-15 19:25:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930970","meta_rev_id":"da5654c60fc93e5257a07c02b7f865cd9c5e1650","_number":930970,"virtual_id_number":930970,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-15 19:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-15 17:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-15 17:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 05:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fa4805cd3ea454b0cb6009e4f83ea261d38048d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 02:50:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83aa04a879528cfb9e4872919cbcbb0f3acc18e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 05:14:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63abcc01ac3b4fe9a9c83070c83165b4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c7e7af8b0c734d72bd1abe6b200a65b3 : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d405bf69b554b0987e741083aac7ecc : SUCCESS in 1h 48m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02afe96a27df4d0791e88e361d72d2db : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/004f5e039d85493da57bcc0e69f4aec6 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6fc58dbd13849fdb62146279563cdff : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/872199b35bdb423482abc60be8e65975 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df686ff5fd664e63a4a37ac04ef064cf : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9aa45abb8dd33f59c716b33e452c5ec58515a6a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-15 17:44:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"638fe56902c942472a7f1bddf939fd0d1609024a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:45:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c56e3ff2b5c7e8c02285d0a3fab1851f54658641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:25:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5304deaaa09e4ce083783f30ea83813c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/74ccb86c9a11460cb194dbea4658f6ec : SUCCESS in 1m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f22f0884140d49f6b0e4e9d1855d1df0 : SUCCESS in 1h 35m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb6ea2fc1b54e64a839bdd3a13b187d : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ed9a6eb6edd45a1a96e83faa734e896 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c543df90e498493695f2e1404a2fdc2a : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d704bba55ed49118b5ca98b84068f4d : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"36b70e1032c6acd965acd20df9e0f1910b2943a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:25:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da5654c60fc93e5257a07c02b7f865cd9c5e1650","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:26:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f784cc77d5a7474daa9ef076903e2f1f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed5021890234ef1b90e2b47a35ac09a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd37e3eb279f7d30bad4d885830aaf0c3ed76212","revisions":{"dd37e3eb279f7d30bad4d885830aaf0c3ed76212":{"kind":"REWORK","_number":1,"created":"2024-10-01 02:50:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/930970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/70/930970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/70/930970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/70/930970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/70/930970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/70/930970/1"}}},"commit":{"parents":[{"commit":"dcfda50d90ad3bb2c09f194c1b75476f2d21478e","subject":"Replace deprecated datetime.utcnow call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/dcfda50d90ad3bb2c09f194c1b75476f2d21478e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 02:50:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 02:50:04.000000000","tz":540},"subject":"Remove unused debtcollector","message":"Remove unused debtcollector\n\nThis library is not used by any code in this repository.\n\nChange-Id: I7df0e750c5e678e5d150382c23c969ca820650b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/dd37e3eb279f7d30bad4d885830aaf0c3ed76212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/dd37e3eb279f7d30bad4d885830aaf0c3ed76212"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
