)]}'
{"id":"openstack%2Fpycadf~951080","triplet_id":"openstack%2Fpycadf~master~I23c1f8b624aefef643deedbf25e4b0ebc6b0f148","project":"openstack/pycadf","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-28 08:38:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23c1f8b624aefef643deedbf25e4b0ebc6b0f148","subject":"add pyproject.toml to support pip 23.1","status":"MERGED","created":"2025-05-28 02:59:32.000000000","updated":"2025-05-28 08:38:47.000000000","submitted":"2025-05-28 08:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951080-pip-23.1-support","meta_rev_id":"fa18de5b35776ce8e3a6731596554b970c74389e","_number":951080,"virtual_id_number":951080,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-28 08:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2025-05-28 07:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2025-05-28 07:15:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 04:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 07:15:13.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"8df72cdc37012c2cd67fa2689d97ca96d9f9a844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-28 02:59:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a3f291cccfa5433b7a8c855fbf46717b6345e72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 04:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee74a7fa6564801975586e900583914\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18acc9d389e144348b79cc325fb07c7a : SUCCESS in 1h 28m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dc0803958534750aff7b8b175d608bf : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c0c933fd9040ff97c36d7101568188 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1203ef0fffa54ff3adb4be4e8ca9a4e3 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6bf6bbb82d647c394de82e9343eb429 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02e2286db53c48b5927fbfe8e46c0f3e : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b08fa7194f4bd78234564840691624 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"93f93a52678a8fa83c0cae17942d1251e949c697","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-28 07:15:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53f88159047d001aca50d34acab4e379fc773e33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 07:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c75e6aaad18ad9407bf623109bd4f81f586f7e7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 08:38:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f775bee60f2438cae5179fe18af83a8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78623665a52240e1a6e3da4abebca713 : SUCCESS in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/064b0bd07e07403e98fa4bb0a0734a66 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a8d2405e19d4704a8e4a0bb1ad0e483 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b10f5b7e237a4d3f924c31fdc7cfbe29 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559eb48f170949bd851b7067728cd8e9 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"048a3a86e8126f81d2a2cc058f75e7ab061c1341","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 08:38:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa18de5b35776ce8e3a6731596554b970c74389e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 08:38:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb1b4772bd84ded830352eedb63b168\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3fef755b3e4c0fae51a0ca6f70f2ad : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af39dfe4754d79002b74cff78613f7a27b3c6aad","revisions":{"af39dfe4754d79002b74cff78613f7a27b3c6aad":{"kind":"REWORK","_number":1,"created":"2025-05-28 02:59:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/951080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/80/951080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/80/951080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/80/951080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/80/951080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/80/951080/1"}}},"commit":{"parents":[{"commit":"d4d5568c977ae2f0912a22478fad3bc39a31ca87","subject":"Merge \"Fix outdated envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/d4d5568c977ae2f0912a22478fad3bc39a31ca87"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-28 02:59:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-28 02:59:23.000000000","tz":540},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I23c1f8b624aefef643deedbf25e4b0ebc6b0f148\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/af39dfe4754d79002b74cff78613f7a27b3c6aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/af39dfe4754d79002b74cff78613f7a27b3c6aad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
