)]}'
{"id":"openstack%2Fpycadf~952366","triplet_id":"openstack%2Fpycadf~master~I49d4394460acd22de6ac5ed22d5230867236d186","project":"openstack/pycadf","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-15 11:41:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-09-15 10:43:02.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I49d4394460acd22de6ac5ed22d5230867236d186","subject":"Add ruff","status":"MERGED","created":"2025-06-11 12:55:27.000000000","updated":"2025-09-15 13:22:49.000000000","submitted":"2025-09-15 13:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952366-pyproject","meta_rev_id":"fb13bd4f10aa3cb9fd162c038eafee251d703bbb","_number":952366,"virtual_id_number":952366,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-15 13:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-15 10:43:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-09-15 11:41:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 14:43:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-18 14:20:29.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2025-09-15 10:43:02.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"689dbbf08b220c1c6f6ce63660c77ebbe89dabb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-11 12:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c2f587b9061fd87f48821abc636a91a647a57cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-11 12:59:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4df15ad59e13978a239d0fc3493a9055cba0b310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 14:43:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eeb2090a76c747b9845303429f36df16\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d81667c2e0e24b55869c5246907b7a08 : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdbf4467f92849df9ef5bdde66ba9fa1 : SUCCESS in 1h 33m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e2cb55f76b48f8a8714ef3d6ea03d7 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51fcf8dbbb5b4705b32b6cb4f84b3780 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48ab1f6c05a54357add4cc934cecdd4c : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/933bfead7eb84dc6b006ff869a0762e3 : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26875406680e4173955ed5240dd9f55c : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bdbb8b58c284b99b6c27de1a5102c70 : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a54baa0eedacad1716d599de83e9f254f3daa115","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-07-18 14:20:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"448b3ba7ee6e44933c9fa73b4e6fce05cc2ee2ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 15:21:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be9ca3407b664d5489b6bdcf485142cd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/acce1d42ce39498a8eeef1841e4e2c8a : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5e83b0fc62842959ed8d377fe71ef2e : SUCCESS in 59m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/737f53cac14c45db82161c145185585f : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc92658ee64b4fbd82eea987437cec59 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c3da0edf0cd471d9077d008b3dd382e : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/182387b6aa594576ac6705ab80cac48c : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65cbdbbee62b467aa622bdc760d6412e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5715d69d45f449e4a51bdcacbb369e43 : FAILURE in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d0b820f18243ce109a8d0186a6c194ea4e39627d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-07-18 15:34:43.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"54f5509c6ad25092e1d804f5772d2f27825c5e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 17:21:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6511818785d43478780968f620fef83\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04248f2bcb814b51981e0cd245e71be3 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcbeefebd64a4ced922c0c5bf8490587 : SUCCESS in 1h 41m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7dc5c3278c749088778719952ab0f51 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/249ab441f4d24d6e844b00a72c167b49 : FAILURE in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ba8fcc55b07419eb99f8e897908e8bd : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf28ec1ec55446e09787a9ef6b50d727 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92c6e569388a40ba866beaeac047480c : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bf735a75f6749c9ab7aa9572ffdb934 : FAILURE in 1m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6fb271d04338e4225050f4157695e6b09467fc17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-12 16:59:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7571babe24c6b66af40c7e2468729582ff1b8c22","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-12 17:00:02.000000000","message":"Topic set to pyproject","accounts_in_message":[],"_revision_number":4},{"id":"019bdfaf9812375d825f116646ee80c732b6285c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 18:47:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66e087664efa4b0eb3dcbadec0791b8e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/61215cb402304e2e94fd2bfb898d585a : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95245de31f99497687bd840be1e828ac : SUCCESS in 1h 43m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a856bef2808c470fa6df7a37bc7273c4 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07ff88c453f4b18a171348c3a423098 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cb4bc947c4641d8ab513dd3a9a59e5f : SUCCESS in 1m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7dd200bdbfa44ef69b087c6ad03ca7ae : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7aba1717285d4016968c6fc26a548132 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed44e17febd742ddb32efe3d7171584f : FAILURE in 2m 02s","accounts_in_message":[],"_revision_number":4},{"id":"26fba2257f500108ce66a0927abd3cfdc5e48e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-15 09:40:19.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"45efd928c0770f20615528accdea414278698ed8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-09-15 10:43:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"37080aaee0de3e98d7651da71af5d80ce66b5ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:32:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2915bd91aaa469e9092535fb99f4112\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42a2f488e42840fda76f3f0b1a8a8c7e : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6c66b95d391436cbd443007feffd189 : SUCCESS in 1h 36m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e321d9dbbe0f4890a77e2b378315b9a4 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe63a5bc93d4db183107acb44b70e02 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27a8b93e6106430f9c8c9492c2a86a55 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da57c4bf08d446319e16b5b04be17684 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cae9382fe24e4465b7ac9159e5781293 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f103c635745548c89c662644b0db3351 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":5},{"id":"859e2542119e3bb91f96889892e07bf0e0e22acf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-15 11:41:58.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e5f9df200e7daae93b456cb9f97ab5ddc32b07dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:42:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fb32088ec5a456b78724555035a6bdd2d91aa101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 13:21:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a124cd04605e41eca9182ea3149bae82\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51f8a750350e4d6dae32a2e93ab7c84c : SUCCESS in 2m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2de9a759359e454281b15ea4424f9120 : SUCCESS in 1h 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ca60e5aab54ed6a50d36cf6b92efce : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3296d7c8cff41ce8714f39488815d28 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af2ad4abd54c4f03b24982ac2408f965 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9622971ce9645d0abc734a9c22724eb : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":5},{"id":"a1e9f62116386cd07636088c03010dcb36484b0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 13:21:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fb13bd4f10aa3cb9fd162c038eafee251d703bbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 13:22:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3fb16ec52244d07a5f6e61e313da611\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456387bab92345f39cc95b7399c00af1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ca15eb31109453f56228f3276b8f88cf67210943","revisions":{"4b698fb75633d93eb0f5bb2244c70fcc259237cc":{"kind":"REWORK","_number":1,"created":"2025-06-11 12:55:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/952366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/66/952366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/66/952366/1"}}},"commit":{"parents":[{"commit":"750932bdb4ce500936a56f6387caf1f388cf4a3d","subject":"Remove cruft from docs config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/750932bdb4ce500936a56f6387caf1f388cf4a3d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 12:47:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 12:55:20.000000000","tz":60},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I49d4394460acd22de6ac5ed22d5230867236d186\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/4b698fb75633d93eb0f5bb2244c70fcc259237cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/4b698fb75633d93eb0f5bb2244c70fcc259237cc"}]},"branch":"refs/heads/master"},"01f52ab995bceba1ed9f79978f9b4bcdecfca841":{"kind":"REWORK","_number":2,"created":"2025-06-11 12:59:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/952366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/66/952366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/66/952366/2"}}},"commit":{"parents":[{"commit":"b636d2833643c1cfea83ae96db4f263848443241","subject":"Remove cruft from docs config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/b636d2833643c1cfea83ae96db4f263848443241"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 12:47:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 12:59:45.000000000","tz":60},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I49d4394460acd22de6ac5ed22d5230867236d186\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/01f52ab995bceba1ed9f79978f9b4bcdecfca841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/01f52ab995bceba1ed9f79978f9b4bcdecfca841"}]},"branch":"refs/heads/master"},"518bc30127df68a0f364fe83b2327225d316f7a1":{"kind":"REWORK","_number":3,"created":"2025-07-18 15:34:43.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/66/952366/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/66/952366/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/66/952366/3"}}},"commit":{"parents":[{"commit":"b636d2833643c1cfea83ae96db4f263848443241","subject":"Remove cruft from docs config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/b636d2833643c1cfea83ae96db4f263848443241"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 12:47:26.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-07-18 15:34:42.000000000","tz":0},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I49d4394460acd22de6ac5ed22d5230867236d186\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/518bc30127df68a0f364fe83b2327225d316f7a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/518bc30127df68a0f364fe83b2327225d316f7a1"}]},"branch":"refs/heads/master"},"fb0a6fe2ed4b55c07a5d8759f387ac77acdf16db":{"kind":"REWORK","_number":4,"created":"2025-09-12 16:59:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/952366/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/66/952366/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/66/952366/4"}}},"commit":{"parents":[{"commit":"3a948eaeaf2b698eb259bf5cea2956334590c8b9","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/3a948eaeaf2b698eb259bf5cea2956334590c8b9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 12:47:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-12 16:59:16.000000000","tz":60},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I49d4394460acd22de6ac5ed22d5230867236d186\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/fb0a6fe2ed4b55c07a5d8759f387ac77acdf16db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/fb0a6fe2ed4b55c07a5d8759f387ac77acdf16db"}]},"branch":"refs/heads/master"},"ca15eb31109453f56228f3276b8f88cf67210943":{"kind":"REWORK","_number":5,"created":"2025-09-15 09:40:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/952366/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pycadf","ref":"refs/changes/66/952366/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pycadf refs/changes/66/952366/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pycadf refs/changes/66/952366/5"}}},"commit":{"parents":[{"commit":"3a948eaeaf2b698eb259bf5cea2956334590c8b9","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/3a948eaeaf2b698eb259bf5cea2956334590c8b9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 12:47:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-15 09:40:19.000000000","tz":0},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I49d4394460acd22de6ac5ed22d5230867236d186\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/ca15eb31109453f56228f3276b8f88cf67210943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pycadf/commit/ca15eb31109453f56228f3276b8f88cf67210943"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
