)]}'
{"id":"openstack%2Fpyeclib~790860","triplet_id":"openstack%2Fpyeclib~master~I3c2518c5c9bfc0bf6ce58cdc82ddafb45c4c5d76","project":"openstack/pyeclib","branch":"master","topic":"py310","hashtags":[],"change_id":"I3c2518c5c9bfc0bf6ce58cdc82ddafb45c4c5d76","subject":"Switch test runner from nose to pytest","status":"MERGED","created":"2021-05-11 23:03:02.000000000","updated":"2021-12-03 20:05:41.000000000","submitted":"2021-12-03 20:05:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790860-py310","meta_rev_id":"c4dd5481eedda745a118035e1ee3af17fb6f49b3","_number":790860,"virtual_id_number":790860,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-03 20:05:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-03 19:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-03 19:49:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 23:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ef5fd33a1dec06dff583196af3f98d6ba7207ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-11 23:03:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a881893e77207fded295815e69e4029d9916e9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 23:17:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce9f036216e4112aaf16aa138127809 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f8425444c454dd7b101cd894d4a0e3a : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c4046357b3c474285796cd2b31fb908 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a0cd033d3f94c9898aff46cde3b0e4c : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5e1a439df1b479a962aa593e05eeb37 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73befdacfca1498bba705b2b9b281bfb : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ded33273a474cdf895a56f13454932e : SUCCESS in 6m 05s\n- pyeclib-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/2d4b47f044af4af0bf549f41ba56213b : SUCCESS in 5m 04s\n- pyeclib-tox-py36-centos-8 https://zuul.opendev.org/t/openstack/build/e27567aae6924c84850ae9452c48de9e : SUCCESS in 4m 57s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/c01fa495014e47f3a823ff0190323098 : SUCCESS in 12m 16s\n- libec-pyeclib-unit-centos-7 https://zuul.opendev.org/t/openstack/build/b0ce453eecee4576a008cedbda1f24f8 : SUCCESS in 10m 46s\n- libec-pyeclib-unit-centos-8 https://zuul.opendev.org/t/openstack/build/ed68baff647a41aca74740134738c464 : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"71e283d5c25e20c90c209ac976a5bc60ddcd6687","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-03 19:49:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cee614b0b4161330771acaab92fa183409fc2ba5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 19:49:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"017dac1bdd92bb0f34873c90447265aaf8158d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:05:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bfd9bd7529740be96bf22c60f30d72f : SUCCESS in 2m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee233a7eb5a24d87b742c9b8a24cc6ef : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/83ffc476a99947e8b926eb4c06bc27ba : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d077548003e4f0bb39e2e146a2dd04a : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7be25996c99e4e6a9678d0ba2a83111e : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8f45a27a2af4f18b9ad007ab1b289a3 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e10d7766174c4cb3b845c657442a7df9 : SUCCESS in 4m 02s\n- pyeclib-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/2955308e8d9b4a0cad72c8fd964527ea : SUCCESS in 4m 03s\n- pyeclib-tox-py36-centos-8 https://zuul.opendev.org/t/openstack/build/ee8db7014b8b467fb4df1496f4222b7d : SUCCESS in 4m 48s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/3db615089d2740ad8fdb956ff7c515a3 : SUCCESS in 8m 13s\n- libec-pyeclib-unit-centos-7 https://zuul.opendev.org/t/openstack/build/09287abf92434dbead4e3db8ae0512bd : SUCCESS in 11m 07s\n- libec-pyeclib-unit-centos-8 https://zuul.opendev.org/t/openstack/build/770079aa02be4ef1a6f6edd3bd1d5517 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c4dd5481eedda745a118035e1ee3af17fb6f49b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:05:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a21dc9f46fc59d8e9f720c5c003b4b65a1c124b1","revisions":{"a21dc9f46fc59d8e9f720c5c003b4b65a1c124b1":{"kind":"REWORK","_number":1,"created":"2021-05-11 23:03:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/790860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/60/790860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/60/790860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/60/790860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/60/790860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/60/790860/1"}}},"commit":{"parents":[{"commit":"a7bf798b1000c037bcf9f54f4deb875aba08ce51","subject":"Use Py_ssize_t when calling PyArg_Parse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/a7bf798b1000c037bcf9f54f4deb875aba08ce51"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-11 23:02:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-11 23:02:42.000000000","tz":-420},"subject":"Switch test runner from nose to pytest","message":"Switch test runner from nose to pytest\n\nChange-Id: I3c2518c5c9bfc0bf6ce58cdc82ddafb45c4c5d76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/a21dc9f46fc59d8e9f720c5c003b4b65a1c124b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/a21dc9f46fc59d8e9f720c5c003b4b65a1c124b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
