)]}'
{"id":"openstack%2Fpyeclib~793985","triplet_id":"openstack%2Fpyeclib~master~I897cefeb2dd1a5f3b097fc89acc9d336f5f71749","project":"openstack/pyeclib","branch":"master","hashtags":[],"change_id":"I897cefeb2dd1a5f3b097fc89acc9d336f5f71749","subject":"Switch IRC references from freenode to OFTC","status":"MERGED","created":"2021-06-01 15:17:16.000000000","updated":"2021-10-21 05:18:02.000000000","submitted":"2021-10-21 05:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793985","meta_rev_id":"0cfd2299142b27dc2002c0dbbb4521bf1b97d7f4","_number":793985,"virtual_id_number":793985,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 05:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-21 05:05:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-21 05:05:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-01 15:28:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 08:35:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"2f4e02760f36540eb53dd5526f7e6057ba0cf34a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-01 15:17:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3757c86cbfdd5d11367a0c9456b710df28d68bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:28:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1185a867fd1e4f10bec2d0e81ca70365 : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac5041b43134f00a76e71c980fb8c5d : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ecfb8433c2dd051392bb7584631fec39ed5d7108","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-02 08:35:07.000000000","message":"Patch Set 1:\n\nHello,\n\nHere is my reply to your IRC message on OFTC #openstack-release:\n\n\u003e so i think this is the right place to ask: why don\u0027t we include a bindep role as part of the test-release-openstack job? if we\u0027re hoping to build wheels, it seems like there\u0027s a non-trivial chance we\u0027ll need to compile against some C library...    \n\u003e for context, it\u0027s preventing pyeclib from being able to update its README to change IRC networks in https://review.opendev.org/c/openstack/pyeclib/+/793985   \n\nI don\u0027t know why we don\u0027t we include the bindep role as part of the test-release-openstack.\n\nAt first glance, I guess you are right, that could be a good idea.\nHowever this is not really a release topic but more an infra topic because this config is maintained in project-config.\n\nI think you are right, that can\u0027t hurt. As you said that could help to compile with C libs.\n\nPlease ask more informations to the infra team about this topic. I think that it could be interesting to start a ML thread to better sync the teams (release, infra, pyeclib).\n\nHopefully it will help you.","accounts_in_message":[],"_revision_number":1},{"id":"d614c9e6ad308753d3248c479d66f212dcc24b89","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-21 05:05:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed52ffd4a6bcba9b89289b77f3426b77b7083a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 05:13:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b741c2c19024c3f96a318e45349698c : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f4e44c63c54fe7b348389fcb2ceede : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9f2cb29205e73266c372110fd9335cddb8c6fc2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 05:14:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3828e3b8fdf73e86ec09a674c373f4da4756d82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 05:18:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b25a1bd2dafc4a02bd7a68aa1b234cb4 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b83574587946edbdb9002a57bf5548 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0cfd2299142b27dc2002c0dbbb4521bf1b97d7f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 05:18:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55a9c980aeb94142cf7adf959ddf31f1a74a1eda","revisions":{"55a9c980aeb94142cf7adf959ddf31f1a74a1eda":{"kind":"REWORK","_number":1,"created":"2021-06-01 15:17:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/793985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/85/793985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/85/793985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/85/793985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/85/793985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/85/793985/1"}}},"commit":{"parents":[{"commit":"f2f83215b67749579c6d49f57b395c2e7928c499","subject":"Fix liberasurecode-git tox target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/f2f83215b67749579c6d49f57b395c2e7928c499"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-01 15:16:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-01 15:17:04.000000000","tz":-420},"subject":"Switch IRC references from freenode to OFTC","message":"Switch IRC references from freenode to OFTC\n\nSee http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022718.html\nand http://lists.opendev.org/pipermail/service-discuss/2021-May/000249.html\nfor more information.\n\nChange-Id: I897cefeb2dd1a5f3b097fc89acc9d336f5f71749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/55a9c980aeb94142cf7adf959ddf31f1a74a1eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/55a9c980aeb94142cf7adf959ddf31f1a74a1eda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
