)]}'
{"id":"openstack%2Fpyeclib~798010","triplet_id":"openstack%2Fpyeclib~master~I7c3eb54c72b3a3b02490a3ad4df78276593e1890","project":"openstack/pyeclib","branch":"master","topic":"rhbz/1967534","hashtags":[],"change_id":"I7c3eb54c72b3a3b02490a3ad4df78276593e1890","subject":"DNM Drop explicit /usr/lib","status":"ABANDONED","created":"2021-06-24 19:34:12.000000000","updated":"2026-01-13 05:42:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0f6e8840e4f4fa8e0e6dca0c5374ce7e8fce98e","_number":798010,"virtual_id_number":798010,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-24 20:06:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 19:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-24 20:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d59abea2a07be56cfb9e3fdcef485060faaa1bd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-24 19:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"313a572c34b5def0a53ac555fbebf289b9e71aee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:49:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/b81f938d4fc74ade891fc93ab6c52104 : SUCCESS in 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"292f20325ccd39162a8e919988c6f53d448037b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 20:06:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a93c6ae6c4d74dea8f04e5ac13755459 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c65533aa4304c75ab22b4caed30dd88 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da920617936a4a28a85df4716ca2f4df : SUCCESS in 6m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a772c8fb9c1d4d97ba04ec3bc02e3de9 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14632e2bf4dd45bcb87868c585004b25 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f4f1a1fba4a41b782d1d31b1d77d98e : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e210eccf9f5b47e1b1653f89d608c4ce : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc67e0a7f8d44606b400062b2a34e8eb : SUCCESS in 6m 48s\n- pyeclib-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/5604cefb14ad4b76933538b53745dc21 : SUCCESS in 6m 30s\n- pyeclib-tox-py36-centos-8 https://zuul.opendev.org/t/openstack/build/cd987811f8d249c58f8d0bf16eeb4586 : SUCCESS in 7m 35s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/8adfada2819c491fa553acf7ba3b062c : SUCCESS in 11m 11s\n- libec-pyeclib-unit-centos-7 https://zuul.opendev.org/t/openstack/build/40d148da2f5a46f485eb1948c71aa629 : FAILURE in 11m 54s\n- libec-pyeclib-unit-centos-8 https://zuul.opendev.org/t/openstack/build/93ba69f629a4416e84b19a68887ad420 : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c0f6e8840e4f4fa8e0e6dca0c5374ce7e8fce98e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-13 05:42:01.000000000","message":"Abandoned\n\nPretty sure this is mooted following the pyproject.toml transition.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e2273d2fa4e9c34fbfc07d075ec8299a941682d","revisions":{"2e2273d2fa4e9c34fbfc07d075ec8299a941682d":{"kind":"REWORK","_number":1,"created":"2021-06-24 19:34:12.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/10/798010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/10/798010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/10/798010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/10/798010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/10/798010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/10/798010/1"}}},"commit":{"parents":[{"commit":"3cc64adebd7136b2d38e065e25164c5d22e564b4","subject":"Merge \"Add a __repr__ for ECPyECLibDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/3cc64adebd7136b2d38e065e25164c5d22e564b4"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-06-24 19:30:32.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-06-24 19:34:12.000000000","tz":-300},"subject":"DNM Drop explicit /usr/lib","message":"DNM Drop explicit /usr/lib\n\nWhen building Fedora RPM, the newest checkers complain that\nthe library contains an explicit RPATH of \"/usr/lib\".\nLooking at the code, isn\u0027t it silly to have a path\nof sys.prefix? It could make sense when prefix was /usr/local,\nbut we supported that one through some explicit checking,\nnow removed.\n\nWIP: let\u0027s see if we can make PyECLib installed in /usr/local\nwithout screwing with explicit arguments to the Extension class.\n\nRelated-To: rhbz#1967534\nChange-Id: I7c3eb54c72b3a3b02490a3ad4df78276593e1890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/2e2273d2fa4e9c34fbfc07d075ec8299a941682d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/2e2273d2fa4e9c34fbfc07d075ec8299a941682d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
