)]}'
{"id":"openstack%2Fpyeclib~906745","triplet_id":"openstack%2Fpyeclib~master~Idcdd6f643330b76328b5cd28c32900eeceacafb8","project":"openstack/pyeclib","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-25 18:28:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idcdd6f643330b76328b5cd28c32900eeceacafb8","subject":"pep8: Use default line width","status":"MERGED","created":"2024-01-25 16:49:15.000000000","updated":"2024-01-25 18:28:21.000000000","submitted":"2024-01-25 18:28:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906745","meta_rev_id":"ef8dc3dc8f61aff39ae5fb8c0e66557e282a9da7","_number":906745,"virtual_id_number":906745,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 18:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-25 17:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-25 17:32:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 17:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-25 17:16:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 17:32:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"2457dd8a528a67bf95e10f04dfa3fc3e089be7ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 16:49:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b6a76178c5664eceb07c9ef3b79296739118746","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 17:01:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1522f15d73af4ad4835bb68de9e41c12\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/e753baf3962c4f65a82dfc105715f6e4 : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67ba6ad8f09974ff4c4e8f6a89efeed66b40c43c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 17:16:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f252e4d5e3e64be1afbfe78a391c8b3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c817aa88e04b39b8fe20f20287eadd : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/175bb822de7c46d6b707d3bb40424fbd : SUCCESS in 3m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d9636997c0149fca8ed66dc64814a30 : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac0a7dc23e0e4cb2a973430473505dda : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69efcfa73bd14ecbacd85cc0f818609d : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97d81a1aa1974779a88513c00ae0ade8 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d8ca611efd64bc099f544726f35214e : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a5d3d830ec04367b2d44309accdf994 : SUCCESS in 3m 54s\n- pyeclib-tox-py311 https://zuul.opendev.org/t/openstack/build/945234984a7049d7ae27cae5d90af2ba : SUCCESS in 3m 23s\n- pyeclib-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/22ec0250edb44cee92911c8083d3eb19 : SUCCESS in 4m 18s\n- pyeclib-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/e606ee4585204facbc4b04bcd733b995 : SUCCESS in 5m 22s\n- pyeclib-tox-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf59686a237444a9b09c82ab6e7e8d04 : SUCCESS in 4m 33s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/deab3ac44c764d43936e87efc8929f01 : SUCCESS in 12m 14s\n- libec-pyeclib-unit-centos-7 https://zuul.opendev.org/t/openstack/build/1d900a48d5354367b11b7b7f6f316d82 : SUCCESS in 15m 27s\n- libec-pyeclib-unit-centos-8-stream https://zuul.opendev.org/t/openstack/build/d87aa6f28f63407b995237432121b81c : SUCCESS in 14m 29s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/810f9464ae74434d8975a6c4a110efc5 : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f5a69978df1bcd425c73312715c2921ee8078e1a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-25 17:32:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e26da688ef4067d0ad799676ddc7aa24423f1f61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 17:32:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1660f3f115f2908d2039d031a60cea4bf9e3cccf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 18:28:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cd4c67c4c28423ab7cbad4382237743\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a5279e80ba4a2b8e7973c082ee1d82 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec19d48f46c24d79bbada72ad50a4744 : SUCCESS in 6m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f3208d444ad4de6ac0f346137064478 : SUCCESS in 13m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41bccd45fd4149c08c9bd9ee294c3b3b : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a8972b7e83146a39692150f98761b34 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e483bc68b4a4f88bf14fe5e1d9e54ee : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26120eb781a64e0783d99e0546d998d3 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f7dc49e0ace47ab9f7ae897484834b9 : SUCCESS in 3m 27s\n- pyeclib-tox-py311 https://zuul.opendev.org/t/openstack/build/9ad9466906654ad2a5a11fbe73364e8e : SUCCESS in 3m 54s\n- pyeclib-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/87fddfe442af490fb670b336e5c3e24d : SUCCESS in 4m 31s\n- pyeclib-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ae00e1bb13f40b883e70f9043a8f4e0 : SUCCESS in 5m 11s\n- pyeclib-tox-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca7cfef728d24bf0b61a4a3a0cd162a0 : SUCCESS in 4m 26s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/cb6779f0ce8f416787276fc2c593fbdb : SUCCESS in 10m 51s\n- libec-pyeclib-unit-centos-7 https://zuul.opendev.org/t/openstack/build/ffc43c942e574508b25da5caa26ac8a9 : SUCCESS in 12m 59s\n- libec-pyeclib-unit-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecd952d61d0b40a0b2e85f0dc1725e40 : SUCCESS in 12m 17s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e108cbf47c14fecacea25c24b15b0a5 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ef8dc3dc8f61aff39ae5fb8c0e66557e282a9da7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 18:28:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bed3b5554083f0e3097de09d1f067c5d4f530651","revisions":{"bed3b5554083f0e3097de09d1f067c5d4f530651":{"kind":"REWORK","_number":1,"created":"2024-01-25 16:49:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/906745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/45/906745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/45/906745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/45/906745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/45/906745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/45/906745/1"}}},"commit":{"parents":[{"commit":"eca7e3605eba547b05ab4f2c2474cce669a92202","subject":"Add release notes to PyPI project URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/eca7e3605eba547b05ab4f2c2474cce669a92202"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 16:45:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 16:49:00.000000000","tz":540},"subject":"pep8: Use default line width","message":"pep8: Use default line width\n\nWe can use the default value by fixing the only one error.\n\ntest/test_pyeclib_api.py:111:80: E501 line too long (83 \u003e 79 characters)\n\nChange-Id: Idcdd6f643330b76328b5cd28c32900eeceacafb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/bed3b5554083f0e3097de09d1f067c5d4f530651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/bed3b5554083f0e3097de09d1f067c5d4f530651"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
