)]}'
{"id":"openstack%2Fpyeclib~932002","triplet_id":"openstack%2Fpyeclib~master~I3e76c3f0a38cd50498e0ddc431f18b60ba134f35","project":"openstack/pyeclib","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-10-21 23:07:05.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-10 15:31:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I3e76c3f0a38cd50498e0ddc431f18b60ba134f35","subject":"Parse wheel metadata","status":"MERGED","created":"2024-10-09 23:23:48.000000000","updated":"2024-10-21 23:07:05.000000000","submitted":"2024-10-21 23:07:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"932002","meta_rev_id":"d52504233c168da2973bda0b1ca62949806397c1","_number":932002,"virtual_id_number":932002,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-21 23:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-21 22:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-21 22:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 03:41:46.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-10-10 06:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3802886ed310f7c476e817b5ce440db09b693b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-09 23:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66a569697328094323b8dd30771d9d308d3965ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-09 23:35:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2fef574a9fa946b03a6258d35a7a2339cc0fe718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-09 23:57:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"15b960714ea5fc9e29a138c405825723d3cdb8c6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-10-10 03:41:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f61f07256d1edd45bb05d375b0b680296423310e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 06:09:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32bb51b7632c4309b949132830a3b19f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd268ac0150491ab3086035353e9732 : SUCCESS in 2m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5dac3897ec564f7bbd07e62a1bd70cbd : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7101e71a14eb4fffb166cf9cbcf6a9c4 : SUCCESS in 4m 28s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/801779689e184609a61726908b530b84 : SUCCESS in 8m 13s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/d1918a76b9ed46d880d41066e581f375 : SUCCESS in 21m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/58927b0e8ab94819a4539fb4bc35b5df : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43295c7be8954bdba76f722fd398a2f8 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6600e6805da485095a9649a43cd76e4 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7240402806a478db30f5a167d950fa3 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1247d8fc9e4b478090cbf50b2488f0a6 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e337c5fed694e268f8270beeaf23d10 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8279f0aa88ac4edf913a219f98af6a8e : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55646612b2cc46d3befe94a30bfd8132 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6274d810402d45a59fd369af729491a7 : SUCCESS in 6m 32s\n- pyeclib-tox-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a86405684b54b619030384446e7a51c : SUCCESS in 4m 01s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/29bb5eb5361940cc8165fea5b4899992 : SUCCESS in 13m 29s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a33d7f9497a4b5d97b5bb79cc30e527 : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":3},{"id":"56d84360215843746cb79ee4e455b9230c3f1c87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 06:14:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eaff455cc1e4fd5a535e7a15208e206\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/94042ef85a2946949975a95af6186aaf : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a91dcceae7777c73ab44f8bd9558093b3b3b607a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-10 15:31:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8ea155c6a3f6a84325c2307e17e9ec5f24322b08","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-21 22:26:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db045644e88c4edb73e630f8a441752f2d707833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 22:26:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a9756dd99d59ab5d909f3a66505964694c6d0f41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 23:07:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf56c0b1b9bb4eaa9a662b01d1ba593b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac4d58f2883487ebe09a4afb992dd66 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1c83f9c5b184307b99d9b153047150f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e8e79853da4b509c4873638b7ff1d4 : SUCCESS in 4m 12s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/92240b1b4b294ea2b9376ab93fae453a : SUCCESS in 9m 13s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/6088e208d8c04901a118fac0bb6c2d86 : SUCCESS in 21m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d13a34336b3c46fd967aba5ad11f60d3 : SUCCESS in 11m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a763ddf90e5e4bdb9c50ebf468e14e01 : SUCCESS in 2m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1448fd3cd42940c2865dcf307b8e5ecd : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d248b0ffda7a40fcb6b1b34c50d739bc : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e2cec1bf1ed4194b42063a094299f35 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba54eae421134e4693e3f86a930212a9 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d10a11070c44c25b336e39ed1602eeb : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29ef2f95da4a4cc0a65555a1181c421c : SUCCESS in 2m 52s\n- pyeclib-tox-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7c11c68fe0c4e04802c94c2ed07c22b : SUCCESS in 3m 54s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/ce64a6e813b04ba0a71024237d41651a : SUCCESS in 7m 12s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/27572887a362430e8485602d5463d076 : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":3},{"id":"d52504233c168da2973bda0b1ca62949806397c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 23:07:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"44833eafe7fc9ee3e413cc107621fa7f6dd7522f","revisions":{"42f1e252020d098cac4a5b54b40dd84588b6f84d":{"kind":"REWORK","_number":1,"created":"2024-10-09 23:23:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/932002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/02/932002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/1"}}},"commit":{"parents":[{"commit":"29f1add6ba835d1b4028bf1a97a4b71e68582cb2","subject":"Build musllinux wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/29f1add6ba835d1b4028bf1a97a4b71e68582cb2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-09 23:23:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-09 23:23:37.000000000","tz":-420},"subject":"Parse wheel metadata","message":"Parse wheel metadata\n\nChange-Id: I3e76c3f0a38cd50498e0ddc431f18b60ba134f35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/42f1e252020d098cac4a5b54b40dd84588b6f84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/42f1e252020d098cac4a5b54b40dd84588b6f84d"}]},"branch":"refs/heads/master"},"d667204ea5c646b9be828742e68dc91e3729de68":{"kind":"REWORK","_number":2,"created":"2024-10-09 23:35:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/932002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/02/932002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/2"}}},"commit":{"parents":[{"commit":"29f1add6ba835d1b4028bf1a97a4b71e68582cb2","subject":"Build musllinux wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/29f1add6ba835d1b4028bf1a97a4b71e68582cb2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-09 23:23:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-09 23:35:43.000000000","tz":0},"subject":"Parse wheel metadata","message":"Parse wheel metadata\n\nChange-Id: I3e76c3f0a38cd50498e0ddc431f18b60ba134f35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/d667204ea5c646b9be828742e68dc91e3729de68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/d667204ea5c646b9be828742e68dc91e3729de68"}]},"branch":"refs/heads/master"},"44833eafe7fc9ee3e413cc107621fa7f6dd7522f":{"kind":"REWORK","_number":3,"created":"2024-10-09 23:57:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/932002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/02/932002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/02/932002/3"}}},"commit":{"parents":[{"commit":"29f1add6ba835d1b4028bf1a97a4b71e68582cb2","subject":"Build musllinux wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/29f1add6ba835d1b4028bf1a97a4b71e68582cb2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-09 23:23:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-09 23:57:25.000000000","tz":0},"subject":"Parse wheel metadata","message":"Parse wheel metadata\n\nChange-Id: I3e76c3f0a38cd50498e0ddc431f18b60ba134f35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/44833eafe7fc9ee3e413cc107621fa7f6dd7522f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/44833eafe7fc9ee3e413cc107621fa7f6dd7522f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
