)]}'
{"id":"openstack%2Fpyeclib~956599","triplet_id":"openstack%2Fpyeclib~master~Idf74137029c8ad70f9e881acf31f8bb0df24edfa","project":"openstack/pyeclib","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-08-27 20:56:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Idf74137029c8ad70f9e881acf31f8bb0df24edfa","subject":"Clean up return/raise handling in check_metadata","status":"MERGED","created":"2025-08-05 19:00:26.000000000","updated":"2025-08-27 21:43:45.000000000","submitted":"2025-08-27 21:43:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"956599","meta_rev_id":"64e20e161ea844cc468c8b17a4fa801fe69d1839","_number":956599,"virtual_id_number":956599,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 21:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-27 20:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-27 20:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-05 19:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-05 19:19:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63b2e8e60872319450ed8cc929868b17c55f9dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-05 19:00:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e504cdf1aa10a880b7d3b1b581504e01846f7eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 19:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e0991765d74064986c5410be6bc829\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/83c9d89ef70a451ebfeecd37c365ba29 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"551b0369162cdeb95095960dd44ae450311b89fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 19:19:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb60df676324fa0812f8c1035caea4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a6818cd9244456a072c4ecced0c59f : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3329294387b43a3b172a294813f8268 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d10efd544cb4e2dadda26eedd74a4db : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df867a78924e4ccf9c28b9d6b728ed67 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/337b2b054c6a4eb28ef3265094bdf2fa : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035814482f934f0880152f42fc8fbee6 : SUCCESS in 6m 06s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/4d895a94c8fa42a7a2929c9f4d1b1515 : SUCCESS in 8m 50s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/e00963574e85469c95bf66f5602fd3cc : SUCCESS in 12m 49s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d463b3ea12748419f58f8326e1e39e3 : SUCCESS in 5m 23s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/d5f2109bcb674ee78ac25303138b47a1 : SUCCESS in 4m 44s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cfc5b3afce64c99b0c934bd2e274ddc : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cee9a3166290150127a03ed018def989444cd2e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-22 17:57:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2d3d76513262800e6981b7243b48257acbe2da20","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-22 18:11:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"997c9014e186cdba15b62d1e49df33f0a7222774","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:21:40.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f4948d28894cc9b6f1654ba5fcc34c\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/95e48d97c81e4b55ad550bdf882fe9f1 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":2},{"id":"688b1c640fed0d09ebed2dbdaf2b173ae99ad8c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:36:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcf4798b71f64414a2ee16ec3acea4da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674d5c2af86544d7894d7347a2f4bca7 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01e1ef4df0374166a52c10be03b11cbd : FAILURE in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f061ea4e1e5493486aa03840cb94529 : FAILURE in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0bb0ac3a91f46e8a2b3935ffa516e40 : FAILURE in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae6d12b550fd4b19bb103202af15d666 : FAILURE in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50013dab5fa0492489957e94eef151fa : SUCCESS in 4m 38s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/4e25648ecb6c4422b630011bc48b7f3f : SUCCESS in 11m 23s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/e600afa404a54c1bb2d43875ba3a54f7 : SUCCESS in 17m 15s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/3aadcae064134ff1a99444bfc13c4a88 : FAILURE in 6m 37s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/6f8b51588f2a44139645793f2c391edd : FAILURE in 11m 01s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/c507357932ae470dbead3c05b64eccbe : FAILURE in 7m 34s","accounts_in_message":[],"_revision_number":2},{"id":"845f0bf17a33653ca8f49edb6c74b52fee5bc00d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-22 20:58:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d62d8d1538e7f23ad6d891d920080e23f4ae36da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 21:53:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca26a32b82874cba8ba8a4ef771775da\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/3f5c95045f6f4cffb98f9f384f23a50d : SUCCESS in 18m 03s","accounts_in_message":[],"_revision_number":3},{"id":"92b0001e43e675f3e1b85454b476e72419cbce59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 22:06:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acc121e5bb7b49009c8a98ace8ebf8d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af45f3a2acf4f02b0fddb45a94ab910 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1e9337d6c3446a69379a8177a0beb52 : SUCCESS in 2m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87ed580ba0e7431ebb408a5bcdb040b9 : SUCCESS in 2m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0e2867541ba4eba9e051ab8f6927e4c : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/caeb2144bfad4216b706090bf6751f52 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43028fd6ac5344b0989f9f4a3ef184b0 : SUCCESS in 3m 30s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/32b997b54b0c432390a8dcea8873ae41 : SUCCESS in 4m 30s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/631aef36c42e4713a8216f638087fbfe : SUCCESS in 16m 03s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/9399ed7c7f624e77abe428dedb67c2dc : SUCCESS in 3m 53s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/64a05bd6848d4b359dbb2b9c3ad6f285 : SUCCESS in 5m 37s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/10857672e90f46ad8a78f4b6248fdc72 : SUCCESS in 10m 30s","accounts_in_message":[],"_revision_number":3},{"id":"25bf6aed2fdaab46207e28c5f9d6134028fa4002","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-27 20:56:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0fdcea9d976058fecb97ce4ca438437275110a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 20:57:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"73763ad90e5ffc130714b8613bd81f4a8aade951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 21:43:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b0e4d7e1e1742028c805ff0ca7a7ede\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c3295148f84a05ad0855afae2df1ea : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a542731682b40b5b3951905fe6bcb36 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bd9538d9cd44e2ea1055a23fc604205 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ead689afd6b4eff97c17d0f03f28ddc : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed867e204234543af3897e315b28130 : SUCCESS in 7m 09s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/3de095ae66ea439391ad901754de5944 : SUCCESS in 9m 18s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/66891c32e404415eb66ae174296848c9 : SUCCESS in 13m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e87c076db584d649bfd2b40e83e6d07 : SUCCESS in 5m 49s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/75946638d2cd42c8b14fb1510eaf1a30 : SUCCESS in 4m 14s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/714efd1f9b17441a8fd140428373ae73 : SUCCESS in 11m 32s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc4cccc5a2fb4b91b56d3791a7002cea : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":3},{"id":"64e20e161ea844cc468c8b17a4fa801fe69d1839","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 21:43:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6e4aa2914f26fe052273da20adda39b06ff4bb6c","revisions":{"0b99aa8b85f86c2362783c5ca3eb82e9e8f379bc":{"kind":"REWORK","_number":1,"created":"2025-08-05 19:00:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/99/956599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/99/956599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/1"}}},"commit":{"parents":[{"commit":"cf4609a5c2e9953cea00c0161b86c631138924aa","subject":"Properly annotate function in check_metadata errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/cf4609a5c2e9953cea00c0161b86c631138924aa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-05 17:26:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-05 19:00:19.000000000","tz":-420},"subject":"Clean up return/raise handling in check_metadata","message":"Clean up return/raise handling in check_metadata\n\nPreviously we could go setting items on a dict that never gets returned\n(nor freed!) or return an empty dict on unknown errors.\n\nChange-Id: Idf74137029c8ad70f9e881acf31f8bb0df24edfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/0b99aa8b85f86c2362783c5ca3eb82e9e8f379bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/0b99aa8b85f86c2362783c5ca3eb82e9e8f379bc"}]},"branch":"refs/heads/master"},"4e971b7ffad5d5ca2fc6bc4a65c0057d03c2e263":{"kind":"REWORK","_number":2,"created":"2025-08-22 17:57:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/99/956599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/99/956599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/2"}}},"commit":{"parents":[{"commit":"cf4609a5c2e9953cea00c0161b86c631138924aa","subject":"Properly annotate function in check_metadata errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/cf4609a5c2e9953cea00c0161b86c631138924aa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-05 17:26:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-22 17:52:14.000000000","tz":-420},"subject":"Clean up return/raise handling in check_metadata","message":"Clean up return/raise handling in check_metadata\n\nPreviously we could go setting items on a dict that never gets returned\n(nor freed!) or return an empty dict on unknown errors.\n\nDrive-by: Check that PyDict_New() succeeded; return a -EBADCHKSUM status\nwhen there\u0027s a bad checksum.\n\nChange-Id: Idf74137029c8ad70f9e881acf31f8bb0df24edfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/4e971b7ffad5d5ca2fc6bc4a65c0057d03c2e263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/4e971b7ffad5d5ca2fc6bc4a65c0057d03c2e263"}]},"branch":"refs/heads/master"},"6e4aa2914f26fe052273da20adda39b06ff4bb6c":{"kind":"REWORK","_number":3,"created":"2025-08-22 20:58:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/99/956599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/99/956599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/99/956599/3"}}},"commit":{"parents":[{"commit":"cf4609a5c2e9953cea00c0161b86c631138924aa","subject":"Properly annotate function in check_metadata errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/cf4609a5c2e9953cea00c0161b86c631138924aa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-05 17:26:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-08-22 20:58:33.000000000","tz":-420},"subject":"Clean up return/raise handling in check_metadata","message":"Clean up return/raise handling in check_metadata\n\nPreviously we could go setting items on a dict that never gets returned\n(nor freed!) or return an empty dict on unknown errors.\n\nDrive-by: Check that PyDict_New() succeeded; return a -EBADCHKSUM status\nwhen there\u0027s a bad checksum.\n\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Idf74137029c8ad70f9e881acf31f8bb0df24edfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/6e4aa2914f26fe052273da20adda39b06ff4bb6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/6e4aa2914f26fe052273da20adda39b06ff4bb6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
