)]}'
{"id":"openstack%2Fpyeclib~974263","triplet_id":"openstack%2Fpyeclib~master~Iead3f27b84c98b4ec07cb346f0f2b1a388b98f9a","project":"openstack/pyeclib","branch":"master","topic":"free-threaded","hashtags":[],"change_id":"Iead3f27b84c98b4ec07cb346f0f2b1a388b98f9a","subject":"Turn off GIL for free-threaded builds","status":"MERGED","created":"2026-01-22 01:25:57.000000000","updated":"2026-03-07 21:57:01.000000000","submitted":"2026-03-07 21:57:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974263-free-threaded","meta_rev_id":"9102de437d70ef9b31e44bb4aa6fc06c76e28298","_number":974263,"virtual_id_number":974263,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-07 21:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-07 21:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-07 21:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 01:43:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-22 01:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e97aebe982bd96fb4dffb4811829d87814bb01f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-22 01:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fadca8e3aac7ef4c0184945ed70a39dcb8e94f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 01:43:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5adc1a757c8446faa4cec0049e83feee\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/a8ef088cd9c647fa9942d2e6c3d130ad : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f8cd8cdad98cd0451e5fdb837b25a10cea8c94de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 01:46:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009fa319d7214ca7bac796f7210ebdea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5a6da5e5bf448eadbbcb2055543067 : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11529bd982024e2c94a5abaab432657d : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fde2bfb0f06a4870af46f6e22a03dfa1 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67bf904d941543a5ad686065476da67e : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a7c4c0d012844c592a2d35bcdb91258 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcd72c00acb44288a982248a41a9402 : SUCCESS in 6m 17s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/a748c0b03c1548d0afbca083c3631fc7 : SUCCESS in 9m 14s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/140978f2e9c8496a96b78d98763089e2 : SUCCESS in 18m 29s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c5b202af31b4d8c8cfad260e5061b8e : SUCCESS in 6m 27s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/5eead1dcaa4a4b44b77bf7e131e89077 : SUCCESS in 5m 20s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/21c30d2034dc42e7824b35ae6227c9f8 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"523e2736190110e9f6b95fa6000741639864728e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-02 19:42:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8dcf32cb543c8760a29790a784dd029e29345b26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 20:04:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08b429406e974ec29fd347ff5bedf6e6\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/c4311a58ea754c7e81eeaf81c66ec450 : SUCCESS in 18m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ba79b77121eaf9ff36974dc58a3fc451cc7243ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 20:09:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9fa397470e64a96bd0c26909f49c967\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc394520ce9d4876803eaaa8ef474efc : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e60c4c2d64b4487a8088ac4af11c8321 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/669c4aeb5ebb4bafb1c3cac201ebf604 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6db35bdff58f40ab8bc618f30fcba63e : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21def55a6cdd4f80a3132c167cb2e597 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d26f6cc0d041d5aa278cef30141e1e : SUCCESS in 7m 32s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/ee54183301d0427d9581f6f161182019 : SUCCESS in 5m 52s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/8ab5c128db654fb2bdc39f1fc4ed1c4c : SUCCESS in 15m 17s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb9008129d124afb9a71a27541c76bcd : SUCCESS in 6m 46s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/03d3317e12c34ca4969ef14d8eacacf0 : SUCCESS in 5m 21s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b9ce31de67a45e8ba91c782a0bd5b26 : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c3b4fca9c3d7e9967727c98108f28dc8b5fc1073","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-03 00:15:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbcd79e42cd69c04da9ef0ddb59cfec3bc508b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-07 04:59:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"942d499a1b4942e5c5ecbc2d12d6b04d0c4bf361","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 05:13:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/605515631f514ff5be82ed3ac815a6cb\n\n- libec-pyeclib-unit-arm64 https://zuul.opendev.org/t/openstack/build/f0739db5a7e94ff699892d0baa2bcd66 : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":3},{"id":"26a1a284bd6cae6f688a91afe252d4c9f21c9913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 05:17:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2bf1e5b49d345cf9fa5809583951cfd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffd539241624d599253ea0bac365382 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9907e80559140b3affcdd03b89f7041 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f93fb985f4b44b339dd0fd2bad305c18 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc53c8c722b8453b99731b00cf7f8b86 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed59c6de44e147708d906429e66bbcd3 : SUCCESS in 2m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a135021c64b473e972b8aaec32ff7fe : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3472ad73d3a842fb8fb576fc1702236b : SUCCESS in 6m 56s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/4cc324db02ea43ba9c87515143450ad2 : SUCCESS in 10m 23s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/f42eaef5ba154c17b975113f431d6230 : SUCCESS in 16m 02s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdea80394c274f2db5d7e99df58a9ac0 : SUCCESS in 6m 02s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/fe2735f90173407b993ca3b8dcc3a284 : SUCCESS in 10m 55s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/d975d2aeb2ee4bf19ccd04de4a9a20fd : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":3},{"id":"9602d59d5cafefd7677757d14440e98c409091c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-07 21:35:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05e086960854792fa18266ec39803f536a3a498b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 21:35:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c945075e45dfbb5ed838b6bb8853a217790fb84b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 21:57:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79650206a7454a4b92a7ada081ae6de8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b0df9d345a4412b7866d8e525b3546 : SUCCESS in 1m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f45830878c0c440bacd6cc9b3d4723a9 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f92595f231964e51aa7c4792163671ac : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8028754bf0b41b88cf7271d8f47a2be : SUCCESS in 1m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52a64afc72824901b40d7efe0924dc62 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d71874fe3f0468088a57dd7680b99da : SUCCESS in 6m 16s\n- pyeclib-build-wheels https://zuul.opendev.org/t/openstack/build/33555502c26a4330a524ded87939dff4 : SUCCESS in 5m 42s\n- pyeclib-build-wheels-arm64 https://zuul.opendev.org/t/openstack/build/502c7ed202054949b1cd7117f4effce9 : SUCCESS in 10m 16s\n- pyeclib-tox-py311-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5bb75902cbf4f49bc8c0f361eb99148 : SUCCESS in 5m 02s\n- libec-pyeclib-unit https://zuul.opendev.org/t/openstack/build/dabb1cebbf884e6ca6b4d42fdde94b06 : SUCCESS in 10m 49s\n- libec-pyeclib-unit-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a31a8e0c62d46d3920d621144c2c33b : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9102de437d70ef9b31e44bb4aa6fc06c76e28298","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 21:57:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"54010f152b47f02b23b87da6bb92fa18965f0455","revisions":{"013804f1f492aac8ec5355c3e2e31580bfef2602":{"kind":"REWORK","_number":1,"created":"2026-01-22 01:25:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/974263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/63/974263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/1"}}},"commit":{"parents":[{"commit":"311f3e4950add2ef9966960d28fa1b62542d8d7b","subject":"Merge \"packaging: Resolve some setuptools deprecations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/311f3e4950add2ef9966960d28fa1b62542d8d7b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-22 01:25:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-22 01:25:30.000000000","tz":-480},"subject":"Turn off GIL for free-threaded builds","message":"Turn off GIL for free-threaded builds\n\n... if running with liberasurecode\u003e1.7.1\n\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929324\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929325\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929847\nChange-Id: Iead3f27b84c98b4ec07cb346f0f2b1a388b98f9a\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/013804f1f492aac8ec5355c3e2e31580bfef2602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/013804f1f492aac8ec5355c3e2e31580bfef2602"}]},"branch":"refs/heads/master"},"25390e530f81c6857e5d2bcc7c24662a25272c48":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-02 19:42:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/974263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/63/974263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/2"}}},"commit":{"parents":[{"commit":"00624f44cfcb3aed6d6b25d2d954a5c70680425e","subject":"Update docker base images to supported versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/00624f44cfcb3aed6d6b25d2d954a5c70680425e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-22 01:25:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-02 19:41:39.000000000","tz":-480},"subject":"Turn off GIL for free-threaded builds","message":"Turn off GIL for free-threaded builds\n\n... if running with liberasurecode\u003e1.7.1\n\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929324\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929325\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929847\nChange-Id: Iead3f27b84c98b4ec07cb346f0f2b1a388b98f9a\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/25390e530f81c6857e5d2bcc7c24662a25272c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/25390e530f81c6857e5d2bcc7c24662a25272c48"}]},"branch":"refs/heads/master"},"54010f152b47f02b23b87da6bb92fa18965f0455":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-07 04:59:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/974263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pyeclib","ref":"refs/changes/63/974263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pyeclib refs/changes/63/974263/3"}}},"commit":{"parents":[{"commit":"8b57eaecf6e6af5277466c6b93ed4c55d4e54581","subject":"Update docker base images to supported versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/8b57eaecf6e6af5277466c6b93ed4c55d4e54581"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-22 01:25:30.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-03-07 04:58:17.000000000","tz":-480},"subject":"Turn off GIL for free-threaded builds","message":"Turn off GIL for free-threaded builds\n\n... if running with liberasurecode\u003e1.7.1\n\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929324\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929325\nRelated-Change: https://review.opendev.org/c/openstack/liberasurecode/+/929847\nChange-Id: Iead3f27b84c98b4ec07cb346f0f2b1a388b98f9a\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/54010f152b47f02b23b87da6bb92fa18965f0455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pyeclib/commit/54010f152b47f02b23b87da6bb92fa18965f0455"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
