)]}'
{"id":"openstack%2Fpython-aodhclient~1006668","triplet_id":"openstack%2Fpython-aodhclient~master~Ie7f27eac881d91add893dffc4c6b5af633d3deb8","project":"openstack/python-aodhclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-22 09:56:25.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-09-22 09:17:02.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-09-22 09:56:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7f27eac881d91add893dffc4c6b5af633d3deb8","subject":"Add ruff and autopep8","status":"MERGED","created":"2026-09-22 01:45:11.000000000","updated":"2026-09-22 09:57:36.000000000","submitted":"2026-09-22 09:56:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006668","meta_rev_id":"21f4cc319cbda2bfa04bfcacb356efcffbc9b223","_number":1006668,"virtual_id_number":1006668,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 09:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-09-22 09:14:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2026-09-22 09:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2026-09-22 09:17:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 02:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 09:14:56.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-09-22 09:15:07.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2026-09-22 09:15:07.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"d3a0d9cceaeb0c7eed4d256d6ff435b6438ec654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-22 01:45:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"310b0d9e550c56fa991b4953ae289d414a3a6a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 02:27:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73887c2eba3246e2898089d9d44939fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d5f87c65c524d4d871bf03597625cd7 : SUCCESS in 2m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/759e82ef93114c89a9ecf9ac03c78a9e : SUCCESS in 1m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9264051756d841e4b3f9018b11d55a52 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7f2f25e0dc94d8fb983a92cbca0b0d1 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18a2b2eaf7eb49d2bb6d8eac7ec16887 : SUCCESS in 2m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b09adff476d46a08b68c81815fdbe8f : SUCCESS in 3m 18s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9fb38edff42f4ce48478992551b3ad6e : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c714164f4d461e9c84189bf5be690f : SUCCESS in 2m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a76fb81ebbbe4a62b389a15eb9721ba8 : SUCCESS in 3m 03s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d741dfb62e8340f89b49b95cc7eaeddc : SUCCESS in 5m 07s (non-voting)\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/e29507e951da418ba65ac6366662dd84 : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"75f1ddf2a2558565974ebddb6cd40da7c22d40ab","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-09-22 09:14:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7274e1ace9dc9047c0796f20b039ab06b6f7b89","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-22 09:17:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"862a0f18fb1edda2cf9024fda67830322fa00df3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:17:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41a565daaec9efb724f1ec68e72cc0a170813a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:56:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1caeb475d61455abf2c118a48869890\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a5960a71fccf4e44b2d5f113b5fc1019 : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1dcf22071ac843838af30db4b5e54afb : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/948b62cf47ee4c99a1cf3f25e5528577 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e971bc9477124bc3a2d183e4215be986 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2e56f4e13c34b5fa79f31e4a1484407 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c79f2148111a4faa954a7744a02c64c7 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1048ac59e16423f9a205cc6ace1a4ad : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1715b7974e244f8685f1d3be5975bf0b : SUCCESS in 3m 06s\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ba247f955a246a0b47a27ee8ffab740 : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3230e46a24106546d27fd4e8aa940ab6d968a35c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:56:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21f4cc319cbda2bfa04bfcacb356efcffbc9b223","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:57:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df50568922e741a3b1d1b6f82775dac8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0daedcd44b94c7497e1ee01dbd99968 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/364c8de1a7d848a6a0630ba4108ec97d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0114b738a458ecce8e397911846160b29d25fc01","revisions":{"0114b738a458ecce8e397911846160b29d25fc01":{"kind":"REWORK","_number":1,"created":"2026-09-22 01:45:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/1006668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-aodhclient","ref":"refs/changes/68/1006668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/68/1006668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/68/1006668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/68/1006668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-aodhclient refs/changes/68/1006668/1"}}},"commit":{"parents":[{"commit":"04e633310a5aa84d447ada3df82796a65c90bfab","subject":"Merge \"Deprecate support for threshold alarm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/04e633310a5aa84d447ada3df82796a65c90bfab"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-22 01:43:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-22 01:44:21.000000000","tz":540},"subject":"Add ruff and autopep8","message":"Add ruff and autopep8\n\nImport the settings from ceilometer to replace the existing checks.\n\nChange-Id: Ie7f27eac881d91add893dffc4c6b5af633d3deb8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/0114b738a458ecce8e397911846160b29d25fc01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/0114b738a458ecce8e397911846160b29d25fc01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
