)]}'
{"id":"openstack%2Fpython-aodhclient~1007507","triplet_id":"openstack%2Fpython-aodhclient~master~I581923b2b7aef9a50d9e73690a6878c0a3c894cf","project":"openstack/python-aodhclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-29 10:48:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I581923b2b7aef9a50d9e73690a6878c0a3c894cf","subject":"Declare Python 3.15 support","status":"MERGED","created":"2026-09-25 15:48:40.000000000","updated":"2026-09-29 10:49:25.000000000","submitted":"2026-09-29 10:48:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007507","meta_rev_id":"791f70367895e6199b5dd575baa2429c72701b5c","_number":1007507,"virtual_id_number":1007507,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 10:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-29 09:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-29 09:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 16:15:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-29 09:44:18.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"023fc2fb3f74649be439be9783a2f0b66ceaab3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 15:48:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d936155eda4814a4323816f5673848e9a79687b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 16:15:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5711ffe1f39041f89eecdc87065aab2e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/be817ec24d254752b3f0da80c36b0371 : SUCCESS in 2m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/24767d6e0f0c42ec9d10d61140a894f0 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3211d1294eb24b7eb75f2d81d70ebc7e : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7385ec3c51c94845b40f5c423ad5a36d : SUCCESS in 2m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a30a550648bf42b99d5f66693f85d739 : SUCCESS in 5m 23s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/eb8000be0a5648b59d20ae2e2757849d : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a35f942b074ec4b90fa78f9b3484ce : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1206fbf53a904295af06205ce227f37e : SUCCESS in 3m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/74adb7f7dc514eaf886386199b1f55e6 : SUCCESS in 5m 34s (non-voting)\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbbfcad5a71543b18a316674761163b4 : SUCCESS in 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"761aa33295a820ba391aa73aa9963161e4a0149e","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-09-29 09:44:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"196615a3fa75f8ea6c512671e77a134af47cac04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 09:45:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72f228d1b453742eabd0907fa8e51eaa733fe8e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 10:48:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3029f92885c4600b063d6d542218980\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/551461c1869543eeb6add92054df16cf : SUCCESS in 2m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/215d9e663a504b72b66db5dce2a1308a : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa7cd4ead554e298b7e1e583f020f9f : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ceeb6d8d11d24fcaa4056234f49360f1 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a4955f01885a45dcb41e0333d4d97fa9 : SUCCESS in 6m 15s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/94296587e31442c2999b4feed3f1124e : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce89b401f62f4fc68615812ead15caeb : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b6bf1d3f4114f3d81293768717b5c3b : SUCCESS in 3m 13s\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/c39eb45b9ada43e987e93f809a3231b0 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"15dc9728fb4c1c38c88d47aab220551ffc3b2e6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 10:48:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"791f70367895e6199b5dd575baa2429c72701b5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 10:49:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ac054a7eec04796ae00b536cbe0b7ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0b3a074f40448c91c3562147f441cb : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f09066319c3fffa680f95745af62c1036c92f7d","revisions":{"1f09066319c3fffa680f95745af62c1036c92f7d":{"kind":"REWORK","_number":1,"created":"2026-09-25 15:48:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/1007507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-aodhclient","ref":"refs/changes/07/1007507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/07/1007507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/07/1007507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/07/1007507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-aodhclient refs/changes/07/1007507/1"}}},"commit":{"parents":[{"commit":"9fee389885d102108bcfa4d4ee2fa59e4fa82dec","subject":"Add doc8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/9fee389885d102108bcfa4d4ee2fa59e4fa82dec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 15:48:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 15:48:11.000000000","tz":540},"subject":"Declare Python 3.15 support","message":"Declare Python 3.15 support\n\nopenstack-tox-py315 job is now voting in the -next job template.\n\nChange-Id: I581923b2b7aef9a50d9e73690a6878c0a3c894cf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/1f09066319c3fffa680f95745af62c1036c92f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/1f09066319c3fffa680f95745af62c1036c92f7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
