)]}'
{"id":"openstack%2Fpython-aodhclient~750976","triplet_id":"openstack%2Fpython-aodhclient~stable%2Fvictoria~I54aa5f13388a660b1ec7477251c4da440ea739d4","project":"openstack/python-aodhclient","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I54aa5f13388a660b1ec7477251c4da440ea739d4","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-10 12:41:11.000000000","updated":"2020-09-11 03:27:15.000000000","submitted":"2020-09-11 03:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750976-1599794758662-7e3b2230","meta_rev_id":"56ad13039c3b9105248a90ad1c170f2d4977d9f1","_number":750976,"virtual_id_number":750976,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 03:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-09-11 00:10:22.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-09-11 00:10:22.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 00:10:22.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-09-11 03:25:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b209ee90049fa8a1cc9f6160e3b998aeea61a9a8","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-10 12:41:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5faf892d14f86e7ea5673d33b987ab366886673d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 12:56:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c9a75f5a214ec2ac54b2b7dbea16bb : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1ca8c2241bc4e43a6e6022fb6448c52 : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/331a97d36b8942768fb5b3ae4171e8df : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae78308b8ce46298d6713e4797f3fa9 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9d4debfd251a1254f66b26a29a0377062e6dd071","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-09-11 00:10:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3039754936746a84d853466728fd2817a3cc6d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 00:10:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5291b5e95f0398f26979c83593254944b2e55e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 00:44:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f85d2c3fda94bb889fa0a59df2bf522 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a216e1664a0b4c5cbb8450e9090205c9 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9df2dfb5e5be4d57a430f1e9f58adfba : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b210bcb39041a5af55b9a27c861900 : TIMED_OUT in 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a755e9c527dbf5a802bb8522abfb027a5b554af9","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-09-11 03:05:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7aa78339f68a50192f1b7dfe539fd9745cf5beb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 03:17:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb818bb840f441cb5ded2b5d6668429 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ca0337fdf4e4ac9ac44a85c54a3672e : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9665eedad21f4f43b8ded248e2f84167 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8536d14de14e329152cd8dd514ede0 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e237aa838d82c04be9c99e07530907ac8680524c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 03:17:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d076fa992e5893cc31ef237b1188de14fbd42101","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 03:25:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2048af87f51b58d46526a74cb70191786882ba79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 03:25:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2911e695214783b8a1dae710ab8728 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/707b7399232b4715931dc44bc13d5e15 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdc6111afc044651be0deace095a1453 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac6f1f26dc3b4e26b64fd3e57eacb6c6 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c5528891d3ee1ec1c0c5855c83fcec1cc2812821","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 03:27:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd90115e7144c9faf1c84f1b90f4dbe : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a18470c866823093d5da5c87d38d4aff4e014f6","revisions":{"3a18470c866823093d5da5c87d38d4aff4e014f6":{"kind":"REWORK","_number":1,"created":"2020-09-10 12:41:11.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/76/750976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-aodhclient","ref":"refs/changes/76/750976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/76/750976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/76/750976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/76/750976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-aodhclient refs/changes/76/750976/1"}}},"commit":{"parents":[{"commit":"dff4f153d442e442276eea0d82f87a5d4cb692fe","subject":"Don\u0027t use */* in accept header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/dff4f153d442e442276eea0d82f87a5d4cb692fe"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-10 12:41:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-10 12:41:09.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I54aa5f13388a660b1ec7477251c4da440ea739d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/3a18470c866823093d5da5c87d38d4aff4e014f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/3a18470c866823093d5da5c87d38d4aff4e014f6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
