)]}'
{"id":"openstack%2Fpython-aodhclient~996901","triplet_id":"openstack%2Fpython-aodhclient~master~I241f33c51f9a463fc8d229e04eac65297f23f5d7","project":"openstack/python-aodhclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-13 15:20:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-08-13 12:24:47.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}},"hashtags":[],"change_id":"I241f33c51f9a463fc8d229e04eac65297f23f5d7","subject":"Replace deprecated pyparsing interfaces","status":"MERGED","created":"2026-07-11 17:37:28.000000000","updated":"2026-08-13 16:50:20.000000000","submitted":"2026-08-13 16:49:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996901","meta_rev_id":"87874ec2bedeab73596e7d6a7ddbc610d39950a6","_number":996901,"virtual_id_number":996901,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 16:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":2,"date":"2026-08-13 12:24:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-13 08:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":1,"date":"2026-08-13 12:24:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-11 18:14:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-13 08:30:06.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2026-08-12 13:10:39.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"bf247687906beda74b17737f6db6189e46fa5876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:37:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd7e049e2573775f677c2e0f646921ff53bd8e8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:39:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef096dc9377bc04fd5363553a0e2a88de54a99d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 18:14:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d5cf5c6e4434128ae7b4b58a1105930\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1480d2bb24f402b92101fee3575dc77 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01a954d12fb419bb2a71b3bf6cec640 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e70e4611593d451ba9bef8205f377916 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d292ddb53ab64201baf5b44c2d991878 : SUCCESS in 2m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/faba255e134c4f68a49775508923b2d6 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbacc01fdc44f28ab5744de6d60bae6 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c00621d9cb04eb3b73896a449775ba1 : SUCCESS in 3m 37s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/06e9218aaca64d99930040355621e9cf : SUCCESS in 7m 43s (non-voting)\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/e01c69f5294444838b20171a6c0a7a6e : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f08c4b10610c43edce3365c7f71672e4af50ff92","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-13 08:30:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d8a2b223a61883ec26cdae20ad9b7c547b5a5e2","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-08-13 12:24:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4cd43448d607820806eaf42d9902ff68ce907854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 12:25:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"681d55a35044b39e8aae453cb67d7b096d6a217f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 12:39:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dfeb12d8a0b4b228fb4b78beafe4144\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c8f8ccd929d4ccb8465dbe669539eef : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd8c47f579894ed2989fd4cdebb9619a : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acf651eb6ec24a8ca857f163396e4bdf : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f101fcaf41443df9b4d6fc41a04d68b : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d3d543658241059d8e433634f0c556 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/148d332741904bf7ab0e71ee6f716888 : SUCCESS in 3m 32s\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e84531233b14412b6c2580bc24c0072 : POST_FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":2},{"id":"531aaac3bb4eac6ca8755fed78676f690b51c9aa","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-08-13 12:49:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d73becbf5b9fee152c34f6095e1076f66cf32e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 13:09:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bb32589787e473f845f02f6c0e76fbb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/65c361d03b234216972882294b0b4497 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107d4631aaa2466ea5bc5baf62ca35cb : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b3d268c02bd44cba90e7faa2ad32745 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/838d590aac3e4a6fb6da306539f1e0dd : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4b9fd441b60d4aceaca7975709096874 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e16a5c17a4e48aeb235b73c8902f97f : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c93a931d13a4876bece8d702ef1d73e : SUCCESS in 2m 39s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/8a5788a6ddfa438984e2d291e43f26db : SUCCESS in 4m 26s (non-voting)\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/c37faeb0004346ecbf027841d5d42e93 : POST_FAILURE in 6m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b717b7b69f0f065beb873abc5e9e5fc6c98a0104","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-13 15:20:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2489b964fa21767944e19e6da875481db900b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:05:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33dcc4f80574c2498894ecd19e3cfd2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/87fca1c68c4a4753a8bf17bc1d705963 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6087f68c25db4c768f3e37da8bcc2c0f : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b99cc393cf9246a8b22bba2ab9cdc97b : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91758bfba53540d39fbfb5786a51f30f : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96a719b01dba43aeae528511615db2ff : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f958fb74abb84d65b3d3013df589bd6b : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60685373b9e34183bee2ad847f98dd2a : SUCCESS in 3m 33s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/54b95c3e8e9840f2931d194c0bd0563d : SUCCESS in 5m 58s (non-voting)\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/33866a03fb334cba9d62bc29b3cf9d12 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3dcc1b0995fffa049fd440d3ab3b4bf61c8fe5ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:05:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81e9a2d493ae234ac7f67762cbe044b849f0bd44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:49:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e6fe1463474b47a4e1e3c97cf74b92\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e15345a02cd04cdebfa0168b7109243d : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55705e505b804a7caf2236e0da0f642d : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bc045ea9e82477ead4938689d80c3c5 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/255c461e69ef446ea4a9e0cb9a75ca97 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c69a3951fb4e229d5581f39b8be207 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe38657d34db43a380f19ff83d1559c4 : SUCCESS in 3m 05s\n- aodhclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3af01429f3a45e1a470c56eff09cf2e : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":2},{"id":"cfcf1fd6ad8d2286a2d1db62c3f0c3951f1f3180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:49:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"87874ec2bedeab73596e7d6a7ddbc610d39950a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:50:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb9394b3cfd4633acae9f906a5167dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/214159b1940449749932c299389c3128 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"823d5e3c223324eb65bf909bed15f6932847cd5d","revisions":{"4e99fb9ed99a71fd9cf88c5e67787af2862a218b":{"kind":"REWORK","_number":1,"created":"2026-07-11 17:37:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/996901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-aodhclient","ref":"refs/changes/01/996901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/1"}}},"commit":{"parents":[{"commit":"c145641cef802807ff337bd54b4e3f4450390c41","subject":"Merge \"Remove Python 3.10 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/c145641cef802807ff337bd54b4e3f4450390c41"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:34:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:37:16.000000000","tz":540},"subject":"Replace deprecated pyparsing interfaces","message":"Replace deprecated pyparsing interfaces\n\nThese interfaces were deprecated in pyparsing 3.0.0[1].\n\n[1] https://pyparsing-docs.readthedocs.io/en/latest/whats_new_in_3_0_0.html#api-changes\n* camelCase names have been converted to PEP-8 snake_case names.\n\nChange-Id: I241f33c51f9a463fc8d229e04eac65297f23f5d7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/4e99fb9ed99a71fd9cf88c5e67787af2862a218b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/4e99fb9ed99a71fd9cf88c5e67787af2862a218b"}]},"branch":"refs/heads/master"},"823d5e3c223324eb65bf909bed15f6932847cd5d":{"kind":"REWORK","_number":2,"created":"2026-07-11 17:39:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/996901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-aodhclient","ref":"refs/changes/01/996901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-aodhclient refs/changes/01/996901/2"}}},"commit":{"parents":[{"commit":"c145641cef802807ff337bd54b4e3f4450390c41","subject":"Merge \"Remove Python 3.10 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/c145641cef802807ff337bd54b4e3f4450390c41"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:34:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:39:44.000000000","tz":540},"subject":"Replace deprecated pyparsing interfaces","message":"Replace deprecated pyparsing interfaces\n\nThese interfaces were deprecated in pyparsing 3.0.0[1].\n\n[1] https://pyparsing-docs.readthedocs.io/en/latest/whats_new_in_3_0_0.html#api-changes\n* camelCase names have been converted to PEP-8 snake_case names.\n\nChange-Id: I241f33c51f9a463fc8d229e04eac65297f23f5d7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/823d5e3c223324eb65bf909bed15f6932847cd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-aodhclient/commit/823d5e3c223324eb65bf909bed15f6932847cd5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
