)]}'
{"id":"openstack%2Fpython-barbicanclient~1003405","triplet_id":"openstack%2Fpython-barbicanclient~master~I1c4e59ba72b6279f38fa040825cdc5a071faacce","project":"openstack/python-barbicanclient","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-02 07:19:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1c4e59ba72b6279f38fa040825cdc5a071faacce","subject":"Fix unmaintained dependency on PrettyTable","status":"NEW","created":"2026-09-02 07:02:15.000000000","updated":"2026-09-02 08:51:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"496e458d07406e996f2371f40160f255bca43315","_number":1003405,"virtual_id_number":1003405,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-02 08:51:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 07:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89d8deec43a0984f5b408debabae80835dd75e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-02 07:02:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70514e5cb50dbeed456ec6cddc611636c5bfa096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 07:19:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ea62802f84242f1ac9b0e6de47896de\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/971c1569c93b48448e731ae97390d464 : FAILURE in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddba0190be664c0fa2db1111f75cb4c0 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b354c8472a04a02809b92458225067f : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cccbb6a100794df08fb3e3d7bc751b92 : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf6c5da0d6404060bb6366e91730e0f4 : SUCCESS in 4m 09s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/84ddb097e73d4971bfcc1b42e461b282 : FAILURE in 4m 55s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4c79c2678af641008482a836bcde9683 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/134feab8a8614f6d8e6ef0a3183c6a45 : SUCCESS in 5m 45s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/673c1e81c21b48cca97e5f3acd4e9e2f : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"27308a77b550ffa795ec956298b44f6f28636c8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-02 08:13:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"496e458d07406e996f2371f40160f255bca43315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:51:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e45c7a56074ecba8ee5f896bc3763f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b082b7aa92eb422a96584913ff388b14 : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cda9ea3fd504ae9a597c39835765f5f : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80921116edaf40db8493d172a8c4809f : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1db8b480c0604bb7bc63d128c152d15b : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc823611383e4a2dbcef2bcd1ffeb1a4 : SUCCESS in 3m 38s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/6851e9efef3e47899ec9858d25a56c81 : FAILURE in 5m 03s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/84cd76431ad24596ba2fa58099c81b76 : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec6c274e92e4705a901a2aa21a86e62 : SUCCESS in 7m 25s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/95e281b8cebf49c58cc611d3bb5528e1 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"688dbf9f646d03e64cd9923466a4109aea712e60","revisions":{"28a20fe4f86f0ca5c8920e960c6f92310d817d1a":{"kind":"REWORK","_number":1,"created":"2026-09-02 07:02:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/1003405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/05/1003405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/1"}}},"commit":{"parents":[{"commit":"3e222f030038ee4e81645be6c4363fde1f64f6e3","subject":"Merge \"Delay string interpolations at logging calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/3e222f030038ee4e81645be6c4363fde1f64f6e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-02 07:01:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-02 07:01:44.000000000","tz":540},"subject":"Fix unmaintained dependency on PrettyTable","message":"Fix unmaintained dependency on PrettyTable\n\nThe library is imported directly but was missing from requirements.\n\nChange-Id: I1c4e59ba72b6279f38fa040825cdc5a071faacce\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/28a20fe4f86f0ca5c8920e960c6f92310d817d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/28a20fe4f86f0ca5c8920e960c6f92310d817d1a"}]},"branch":"refs/heads/master"},"688dbf9f646d03e64cd9923466a4109aea712e60":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-02 08:13:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/1003405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/05/1003405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/05/1003405/2"}}},"commit":{"parents":[{"commit":"7dddce66a6268d05a18d012809d2f0e5e445b3c3","subject":"Add minimum pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/7dddce66a6268d05a18d012809d2f0e5e445b3c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-02 07:01:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-02 17:13:24.000000000","tz":540},"subject":"Fix unmaintained dependency on PrettyTable","message":"Fix unmaintained dependency on PrettyTable\n\nThe library is imported directly but was missing from requirements.\n\nChange-Id: I1c4e59ba72b6279f38fa040825cdc5a071faacce\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/688dbf9f646d03e64cd9923466a4109aea712e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/688dbf9f646d03e64cd9923466a4109aea712e60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
