)]}'
{"id":"openstack%2Fpython-barbicanclient~1003416","triplet_id":"openstack%2Fpython-barbicanclient~master~Id2e412a0e40bc0a473b7b00851774bebf2c6a7e5","project":"openstack/python-barbicanclient","branch":"master","hashtags":[],"change_id":"Id2e412a0e40bc0a473b7b00851774bebf2c6a7e5","subject":"Add minimum pyproject.toml","status":"NEW","created":"2026-09-02 08:13:42.000000000","updated":"2026-09-02 08:38:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41f225479e24c2657ecd3e451fdca6f4aa318b07","_number":1003416,"virtual_id_number":1003416,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-02 08:38:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 08:38:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2487756032667b176350fd7cb272669e0ec11666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-02 08:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41f225479e24c2657ecd3e451fdca6f4aa318b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:38:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bf7a90c700b4b8bb018d49ad9191c8d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26fdec727be14e1dbef27a76a88cc000 : SUCCESS in 1m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e2fdfdcb533e4633991dc03b6f5a290e : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8525c5770043458b9829f8a0b852edbc : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a06ee45cd3694f23b67ab5afc461cb7f : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdedd07f005b43eb95bae80555b22448 : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd33857a56d942c1a369d9539c1ac2d4 : SUCCESS in 4m 13s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/eca73be042b04d10b0fedcaa5e491289 : FAILURE in 4m 53s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b5f3af70b32d438ab2cfb2d161a0fbad : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808ef74844864ce7884058e34770c9dd : SUCCESS in 7m 37s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/9809aa8f7bac4a9188cedca06e88b4a2 : SUCCESS in 23m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dddce66a6268d05a18d012809d2f0e5e445b3c3","revisions":{"7dddce66a6268d05a18d012809d2f0e5e445b3c3":{"kind":"REWORK","_number":1,"created":"2026-09-02 08:13:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/1003416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/16/1003416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/16/1003416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/16/1003416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/16/1003416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/16/1003416/1"}}},"commit":{"parents":[{"commit":"3e222f030038ee4e81645be6c4363fde1f64f6e3","subject":"Merge \"Delay string interpolations at logging calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/3e222f030038ee4e81645be6c4363fde1f64f6e3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-02 17:13:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-02 17:13:24.000000000","tz":540},"subject":"Add minimum pyproject.toml","message":"Add minimum pyproject.toml\n\nIt\u0027s now required by the requirements-check job.\n\nChange-Id: Id2e412a0e40bc0a473b7b00851774bebf2c6a7e5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/7dddce66a6268d05a18d012809d2f0e5e445b3c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/7dddce66a6268d05a18d012809d2f0e5e445b3c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
