)]}'
{"id":"openstack%2Fpython-barbicanclient~1005707","triplet_id":"openstack%2Fpython-barbicanclient~master~I682d5b25e3b57b790dc42553103cd516f4be7239","project":"openstack/python-barbicanclient","branch":"master","topic":"topic/payload-content-type","attention_set":{},"removed_from_attention_set":{"39444":{"account":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"last_update":"2026-09-16 09:07:42.000000000","reason":"\u003cGERRIT_ACCOUNT_39444\u003e replied on the change","reason_account":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"}}},"hashtags":[],"change_id":"I682d5b25e3b57b790dc42553103cd516f4be7239","subject":"Use hyphenated payload content type CLI option","status":"NEW","created":"2026-09-15 10:20:23.000000000","updated":"2026-09-16 09:07:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a23cb632c7ae62f18d327d959b868b7d836940ac","_number":1005707,"virtual_id_number":1005707,"owner":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 06:59:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 10:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"565a8fd87e9664bc46205aac18d643a8ba52d85b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"date":"2026-09-15 10:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"280ba03bb8267be005a74d8bb72d1a2b47255762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:58:46.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/312f12bc53694fbda96b696e4734f4d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8777bf0a67f34340af4821f517e9d1f4 : FAILURE in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6bbfe3fd45a47379742460592740dad : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42e1cb3c644f4af7918ef13d6c9f5a13 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/999d4c8fb7f0464daca529c1fef73a0e : SUCCESS in 8m 04s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/420fb0e7ffc24404b55c5ec9345f6103 : FAILURE in 3m 16s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7059b1e8ab7a4c61b06b8ac4604bdc08 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f133c523e0a4fb6ac5642e0b6a653db : SUCCESS in 9m 44s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/efdf0ac5a2a642bdb080aff9a4eea3b6 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bd6b7bac4bec4e45faf33b0526d313b3db22e974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"date":"2026-09-15 11:09:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"64b5ba11847e3cca6a0ccc7c9d8c07e73b148aa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:52:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8253563ebac7464180ab2a49933db3ba\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/09e84a9a49b249f9b8d13ce2ad0c366c : FAILURE in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d804f8b8bffd4c9ea933070ad296ac8b : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97cb0afe8b1948f793fd42d3591780eb : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/088cc4fd84f247e2b78b12ad4b6948db : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8757e6c086c94e5694d01db2110f6c85 : SUCCESS in 7m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8cb93968d730474c8b91da35ae48b8d4 : SUCCESS in 6m 58s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b74fb2d9b1e046b784314ca3d000bbc9 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abad66e8f38e4b1590267b4e728f28c5 : SUCCESS in 6m 50s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/2361393bc63b4298b02148e7a9aa3821 : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5ab1a7ce2d9d6b09937830107e1d8b06ead27e5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"date":"2026-09-16 05:28:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"adf14a84a6ad1a5defb22b9fe56b230a807c4d95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 06:07:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c97a8ab49cc4ae7a905d0de3df274df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f246656322eb435082a1d006cf8222eb : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61314421bc344b5593eea8fd3fecad50 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ba9e0ef54364990bbc32126bedfb633 : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07d3bd50ce874e8eb8b5d3f2f08123dd : SUCCESS in 6m 11s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8b05b3c9ac9248f8b4c6bfb3766d86c6 : FAILURE in 4m 59s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4af39e5d6a3740ed97eaf3466e5cfea1 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583de5150f7449ca995db9ae802b7d73 : SUCCESS in 8m 05s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/85da195f63d24a9daae5297bb944b835 : SUCCESS in 33m 25s","accounts_in_message":[],"_revision_number":3},{"id":"250f0bdc8219adf71482ca71ee1de5adce538a92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"date":"2026-09-16 06:25:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"669c19b765475f56cc5c41ca1dddc1bec9d5bf0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 06:59:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865a38fb8629467691ca9ee96a59cb72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7789bc97c9f94463a82ef87d5a9c4dc7 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b345a9416ff41d49abdb5306277bda4 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76e7edd0bf95467589b759f1d4494e85 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3dc26f6a59147ca948ffe3ff1ce8cb5 : SUCCESS in 5m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1a67722a2726481a96103785c47582b0 : SUCCESS in 6m 32s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d3bd5317018844f8820b31831690761b : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7a5d5171df44c594fceb8be9a64480 : SUCCESS in 6m 43s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/495e9fb0cdb74dcf85e49bc6cc22dc82 : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":4},{"id":"a23cb632c7ae62f18d327d959b868b7d836940ac","author":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"date":"2026-09-16 09:07:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cda4f624d855510b96589ca4a33b6b425707f866","revisions":{"5a1bf5005e0a9a8c7633af3f6716c0b55aee8011":{"kind":"REWORK","_number":1,"created":"2026-09-15 10:20:23.000000000","uploader":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"ref":"refs/changes/07/1005707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/07/1005707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/1"}}},"commit":{"parents":[{"commit":"3e222f030038ee4e81645be6c4363fde1f64f6e3","subject":"Merge \"Delay string interpolations at logging calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/3e222f030038ee4e81645be6c4363fde1f64f6e3"}]}],"author":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-15 10:18:36.000000000","tz":300},"committer":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-15 10:19:59.000000000","tz":300},"subject":"Use hyphenated payload content type CLI option","message":"Use hyphenated payload content type CLI option\n\nChange-Id: I682d5b25e3b57b790dc42553103cd516f4be7239\nSigned-off-by: Askar Sabyrov \u003ckyrkaz@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/5a1bf5005e0a9a8c7633af3f6716c0b55aee8011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/5a1bf5005e0a9a8c7633af3f6716c0b55aee8011"}]},"branch":"refs/heads/master"},"596096d14f66dc8ed9d0defaa928498f4ac9ff6e":{"kind":"REWORK","_number":2,"created":"2026-09-15 11:09:42.000000000","uploader":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"ref":"refs/changes/07/1005707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/07/1005707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/2"}}},"commit":{"parents":[{"commit":"3e222f030038ee4e81645be6c4363fde1f64f6e3","subject":"Merge \"Delay string interpolations at logging calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/3e222f030038ee4e81645be6c4363fde1f64f6e3"}]}],"author":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-15 10:18:36.000000000","tz":300},"committer":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-15 11:09:00.000000000","tz":300},"subject":"Use hyphenated payload content type CLI option","message":"Use hyphenated payload content type CLI option\n\nChange-Id: I682d5b25e3b57b790dc42553103cd516f4be7239\nSigned-off-by: Askar Sabyrov \u003ckyrkaz@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/596096d14f66dc8ed9d0defaa928498f4ac9ff6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/596096d14f66dc8ed9d0defaa928498f4ac9ff6e"}]},"branch":"refs/heads/master"},"4f7d85c12e393b7992e3c5255185b05a501b58d2":{"kind":"REWORK","_number":3,"created":"2026-09-16 05:28:23.000000000","uploader":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"ref":"refs/changes/07/1005707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/07/1005707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/3"}}},"commit":{"parents":[{"commit":"3e222f030038ee4e81645be6c4363fde1f64f6e3","subject":"Merge \"Delay string interpolations at logging calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/3e222f030038ee4e81645be6c4363fde1f64f6e3"}]}],"author":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-15 10:18:36.000000000","tz":300},"committer":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-16 05:26:53.000000000","tz":300},"subject":"Use hyphenated payload content type CLI option","message":"Use hyphenated payload content type CLI option\n\nChange-Id: I682d5b25e3b57b790dc42553103cd516f4be7239\nSigned-off-by: Askar Sabyrov \u003ckyrkaz@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/4f7d85c12e393b7992e3c5255185b05a501b58d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/4f7d85c12e393b7992e3c5255185b05a501b58d2"}]},"branch":"refs/heads/master"},"cda4f624d855510b96589ca4a33b6b425707f866":{"kind":"REWORK","_number":4,"created":"2026-09-16 06:25:41.000000000","uploader":{"_account_id":39444,"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","username":"askarsabyrov"},"ref":"refs/changes/07/1005707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/07/1005707/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/07/1005707/4"}}},"commit":{"parents":[{"commit":"3e222f030038ee4e81645be6c4363fde1f64f6e3","subject":"Merge \"Delay string interpolations at logging calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/3e222f030038ee4e81645be6c4363fde1f64f6e3"}]}],"author":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-15 10:18:36.000000000","tz":300},"committer":{"name":"Askar Sabyrov","email":"kyrkaz@gmail.com","date":"2026-09-16 06:24:53.000000000","tz":300},"subject":"Use hyphenated payload content type CLI option","message":"Use hyphenated payload content type CLI option\n\nChange-Id: I682d5b25e3b57b790dc42553103cd516f4be7239\nSigned-off-by: Askar Sabyrov \u003ckyrkaz@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/cda4f624d855510b96589ca4a33b6b425707f866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/cda4f624d855510b96589ca4a33b6b425707f866"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
