)]}'
{"id":"openstack%2Fpython-barbicanclient~993265","triplet_id":"openstack%2Fpython-barbicanclient~master~I3ef37c0f5b9aae7dc22af285aedba0499218e684","project":"openstack/python-barbicanclient","branch":"master","attention_set":{},"removed_from_attention_set":{"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2026-07-24 15:01:55.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2026-07-24 14:10:57.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-24 15:01:55.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2026-07-24 14:11:27.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2026-07-24 14:11:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9914\u003e replied on the change","reason_account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}}},"hashtags":[],"change_id":"I3ef37c0f5b9aae7dc22af285aedba0499218e684","subject":"Declare Python 3.13 support","status":"MERGED","created":"2026-06-14 15:55:37.000000000","updated":"2026-07-24 15:03:29.000000000","submitted":"2026-07-24 15:01:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993265","meta_rev_id":"8430e95ef34b279545065a2cde3855119c761070","_number":993265,"virtual_id_number":993265,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-24 15:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2026-07-24 14:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2026-07-24 14:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":2,"date":"2026-07-24 14:11:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2026-07-24 14:11:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":1,"date":"2026-07-24 14:11:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-14 16:34:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 03:00:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2026-07-22 03:00:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2026-07-22 03:00:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2026-07-22 03:00:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"}],"messages":[{"id":"f69b9fb67470c50e7e07bfff29766bef84b362bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-14 15:55:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89dd2ba4a9850cfbe295ebcb9f1649e4ec273d9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 16:34:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b990a0ed7e4b1183a4bb037388a01c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d29b57f6d3ab4794b7c3921aae36c931 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc916a5e00c04806a74e5cbdc9db46d3 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27bbcb85d9e140428d169d5fa728d967 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa2d77daa6344644aca04f41421e1bb2 : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e6592dd35f646aa9c1eb4a60bcb1b1e : SUCCESS in 5m 41s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0c86b2bec46b4c2aadf584a012203907 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f44aa9d45c94c2e815763e63c79247a : SUCCESS in 5m 36s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/3da8201b69064bdcac2f3fe2a8e7c622 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f2ba6ab70725d24ac182f0e1a39bf1bcc36794da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-22 03:02:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7038815e8a7b3049217952221c7f0a154f0428cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 03:47:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff7abc698f6b4bada35310f837504118\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b93b77e954484fbe9c7d282093f7614e : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b313b571345f480a8d014317917fb914 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d514ee8829e1460aa2c44fe5d165d78f : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6845708c72dd4b55b6d4b1c4e0b419fb : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e6b46e7869b4d61b5f09347dae81e57 : SUCCESS in 8m 23s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d2d3fe63a1e743029e12778f9be67fa7 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73114a2a58d34c95a95371a42fdc81f2 : SUCCESS in 4m 45s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/46e1bdfcf57b4da38aff76420a293b34 : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":2},{"id":"703d10629c67e640b7161a7210fac4ee4cf8a6f8","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2026-07-24 14:10:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c4306a2d851dfd7df7ae7e5c1bd41dedc418bc3","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2026-07-24 14:11:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e2519546d2d87def6674848541367d36bbe8bcce","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-07-24 14:11:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01ce70c217dc274fb0b26cfadbca078ff4008c3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 14:11:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba4e1c406023bde6496f1600e3822f496fd55c38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 15:01:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/215359e0bd844c249972bdc8a0abda90\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a9e51d174214079b31f4d4522bbb0d4 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae38e22a7a24c5b8d6c4652f28e0ab3 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60ee520554144b5d912a45e1e336bce8 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6b6a4632ca14c45b54641a53f3b6f9c : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e0690df3d645bf8e6991e3215c53ce : SUCCESS in 9m 46s\n- python-barbicanclient-tox-functional https://zuul.opendev.org/t/openstack/build/d0acd76cfd0b4d33899812585bd97889 : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c593be5521017aaef2a02ea9100b057fb8fb0331","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 15:01:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8430e95ef34b279545065a2cde3855119c761070","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 15:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/285f713b914d46169f6b8d94f892a8af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c677bf4a3c6d4ea4ab33c72bbcdc5d43 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"184c12e8af4d0259c4f18e9308658ce0e9d627b9","revisions":{"2fab8a17d03beb58bd5850bcc62383b2b9d8b5f3":{"kind":"REWORK","_number":1,"created":"2026-06-14 15:55:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/993265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/65/993265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/1"}}},"commit":{"parents":[{"commit":"2d03f28f80499a5f24ad30943a5bbfef45758e7d","subject":"Merge \"Replace deprecated datetime.datetime.utcnow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/2d03f28f80499a5f24ad30943a5bbfef45758e7d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:55:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:55:14.000000000","tz":540},"subject":"Declare Python 3.13 support","message":"Declare Python 3.13 support\n\nIt has been tested for multiple cycles.\n\nChange-Id: I3ef37c0f5b9aae7dc22af285aedba0499218e684\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/2fab8a17d03beb58bd5850bcc62383b2b9d8b5f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/2fab8a17d03beb58bd5850bcc62383b2b9d8b5f3"}]},"branch":"refs/heads/master"},"184c12e8af4d0259c4f18e9308658ce0e9d627b9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-22 03:02:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/993265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-barbicanclient","ref":"refs/changes/65/993265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-barbicanclient refs/changes/65/993265/2"}}},"commit":{"parents":[{"commit":"6c77e0d3f31417de16eef472020f0384917fa3ac","subject":"Drop options for old documentation build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/6c77e0d3f31417de16eef472020f0384917fa3ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-14 15:55:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-22 03:01:27.000000000","tz":540},"subject":"Declare Python 3.13 support","message":"Declare Python 3.13 support\n\nIt has been tested for multiple cycles.\n\nChange-Id: I3ef37c0f5b9aae7dc22af285aedba0499218e684\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/184c12e8af4d0259c4f18e9308658ce0e9d627b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-barbicanclient/commit/184c12e8af4d0259c4f18e9308658ce0e9d627b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
