)]}'
{"id":"openstack%2Fpython-blazarclient~681325","triplet_id":"openstack%2Fpython-blazarclient~master~Iad95ebf662fb052000f05aa3d76f76308704fd1d","project":"openstack/python-blazarclient","branch":"master","topic":"floatingip-reservation-update","hashtags":[],"change_id":"Iad95ebf662fb052000f05aa3d76f76308704fd1d","subject":"Support floating IP reservation parameters in lease-update","status":"MERGED","created":"2019-09-10 18:16:25.000000000","updated":"2019-09-13 09:29:15.000000000","submitted":"2019-09-13 09:28:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681325-1568366880575-453d4686","meta_rev_id":"c0e382f3fa2be1a90bc8e0070c11512512113dea","_number":681325,"virtual_id_number":681325,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-09-13 09:28:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":2,"date":"2019-09-12 11:43:10.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":1,"date":"2019-09-13 09:09:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},{"value":0,"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 18:16:48.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2019-09-10 18:16:53.000000000","updated_by":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"reviewer":{"_account_id":13192,"name":"Souville","display_name":"Bertrand Souville","email":"souville@docomolab-euro.com","username":"bertys"},"state":"REVIEWER"},{"updated":"2019-09-10 18:17:00.000000000","updated_by":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"reviewer":{"_account_id":29270,"name":"Jacob Colleran","email":"jakecoll@uchicago.edu","username":"jakecoll2"},"state":"REVIEWER"},{"updated":"2019-09-10 18:17:05.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"},{"updated":"2019-09-12 11:43:10.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-09-13 09:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad37e0e94c8f740ba18f0f7d0fcd863713d1d777","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-10 18:16:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aebd369371217fe1b07c1e01daee4d15d41a87b1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-10 18:21:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9fad373ceddfb89678d357f3307e94ea5cd8d928","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 18:31:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edcdb8534a2747469f5806c69a173f99 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b67f8737e644679e970bfcd1c786ac : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c90c16cbb5249f2995607351b398283 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2321a995caf452ab965e43d8408e4eb : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee3ef489b1e34d9499abc6529c19a2e7 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"ef1ecd8afc0a3e8b10cd19dd621c3072ea48510b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-11 15:50:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2202aeeebc504434f9422ec1748d03c0b79a9bee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:28:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/118d56a21c7a49d8a911bdbad4a78720 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42dbc6e74564fa682d4ab90ca8a71d6 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14f791ec0db742229545b7e108d4b5cd : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f71e9516bd74286b5a3b75cdc9f281f : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec31f3f97b514255b9b84ee4bd04daa8 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7b79b3f2236418881996b119ab2bdc7 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"5bf516778f320e31257c70570a0f8926a982871a","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-12 11:43:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b5d0f66339c35e126b470ffa2fa35fdb15fe0b85","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-13 09:09:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"37097a5012b0095e29c70c7217637b8cb933a4b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:09:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04c0c52e873c295e61d3fae24e2c0714e241c4f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:28:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7f3290301d8c67160616495e1cabb942403cee4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:28:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67f24abe053a451b9876029a0117acd3 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62788c685cfd40f6a34f2eba364b278a : SUCCESS in 2m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c206e03ffb07451398da4bb4e90e93de : SUCCESS in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fca17e7cebdb4044ae0a9b9af0ab8224 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2318ecd85c2c4663ab8fa97bf185832a : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56b61a1e6edc434faa23cae2986ea7e7 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"09825cf7e7d8f1dab97c37b8660b74f0c4ada702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:29:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d44d529b97a4318bc4eb33e5d945c4d : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e98421c9197873d0b3c2c9a9601b7dc0655da72e","revisions":{"dcce76655e00e1a8ab34ce6942ae625f8f5fc0c2":{"kind":"REWORK","_number":1,"created":"2019-09-10 18:16:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/681325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-blazarclient","ref":"refs/changes/25/681325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/1"}}},"commit":{"parents":[{"commit":"91839508335cd5cee3ab1231715914129662db82","subject":"Parse required_floatingips parameter as a JSON array","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/91839508335cd5cee3ab1231715914129662db82"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-10 18:16:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-10 18:16:02.000000000","tz":120},"subject":"Support floating IP reservation parameters in lease-update","message":"Support floating IP reservation parameters in lease-update\n\nThe lease-update command needs to know about all possible keys that can\nbe passed as reservation parameters. Also treat required_floatingips as\na list, like in commit 91839508335cd5cee3ab1231715914129662db82,\n\nChange-Id: Iad95ebf662fb052000f05aa3d76f76308704fd1d\nRelated-Bug: #1843258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/dcce76655e00e1a8ab34ce6942ae625f8f5fc0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/dcce76655e00e1a8ab34ce6942ae625f8f5fc0c2"}]},"branch":"refs/heads/master"},"6a60b3f47b55685256c2a4d0575dd0b7cdc221a9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-10 18:21:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/681325/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-blazarclient","ref":"refs/changes/25/681325/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/2"}}},"commit":{"parents":[{"commit":"91839508335cd5cee3ab1231715914129662db82","subject":"Parse required_floatingips parameter as a JSON array","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/91839508335cd5cee3ab1231715914129662db82"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-10 18:21:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-10 18:21:06.000000000","tz":120},"subject":"Support floating IP reservation parameters in lease-update","message":"Support floating IP reservation parameters in lease-update\n\nThe lease-update command needs to know about all possible keys that can\nbe passed as reservation parameters. Also treat required_floatingips as\na list, like in commit 91839508335cd5cee3ab1231715914129662db82.\n\nChange-Id: Iad95ebf662fb052000f05aa3d76f76308704fd1d\nRelated-Bug: #1843258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/6a60b3f47b55685256c2a4d0575dd0b7cdc221a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/6a60b3f47b55685256c2a4d0575dd0b7cdc221a9"}]},"branch":"refs/heads/master"},"e98421c9197873d0b3c2c9a9601b7dc0655da72e":{"kind":"REWORK","_number":3,"created":"2019-09-11 15:50:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/681325/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-blazarclient","ref":"refs/changes/25/681325/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-blazarclient refs/changes/25/681325/3"}}},"commit":{"parents":[{"commit":"91839508335cd5cee3ab1231715914129662db82","subject":"Parse required_floatingips parameter as a JSON array","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/91839508335cd5cee3ab1231715914129662db82"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-11 15:50:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-11 15:50:43.000000000","tz":120},"subject":"Support floating IP reservation parameters in lease-update","message":"Support floating IP reservation parameters in lease-update\n\nThe lease-update command needs to know about all possible keys that can\nbe passed as reservation parameters. Also treat required_floatingips as\na list, like in commit 91839508335cd5cee3ab1231715914129662db82.\n\nChange-Id: Iad95ebf662fb052000f05aa3d76f76308704fd1d\nRelated-Bug: #1843258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/e98421c9197873d0b3c2c9a9601b7dc0655da72e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-blazarclient/commit/e98421c9197873d0b3c2c9a9601b7dc0655da72e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
