)]}'
{"id":"openstack%2Fpython-cinderclient~986728","triplet_id":"openstack%2Fpython-cinderclient~master~Ia7891c1be82d75d023eef2cc3372cac71847101a","project":"openstack/python-cinderclient","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-10-02 16:41:21.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-10-02 16:41:21.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-10-02 16:41:21.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-10-02 16:41:21.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-10-02 16:41:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7891c1be82d75d023eef2cc3372cac71847101a","subject":"Fix typo","status":"MERGED","created":"2026-04-29 23:27:01.000000000","updated":"2026-10-02 16:42:17.000000000","submitted":"2026-10-02 16:41:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986728","meta_rev_id":"5604f8650ae1c58b2b1901cdd929bf698a16646b","_number":986728,"virtual_id_number":986728,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-02 16:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":2,"date":"2026-10-02 15:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":1,"date":"2026-10-02 15:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 00:55:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-15 12:36:40.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-05-15 15:08:48.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-05-16 19:25:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-08-27 09:44:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-10-02 15:48:21.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"}],"messages":[{"id":"708aac7d7ebbe19c5981d0483fecb23a9fdbfcc3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-29 23:27:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7d4dd94b19c17800f3753326d1192cdc5bf71f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-29 23:27:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4fdbd22e16547ab71bd6a9a0402b5d037793dc19","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-29 23:27:47.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"3c1d6ea2ebc3885fcdd6abd2f1d04467e8ea604d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 00:55:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da51a2e3a684ab4a7027c3a88ce3039\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37f79fd95cc542aab47f05cccd60efb2 : SUCCESS in 1h 26m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2e97b3181b5472aaee2696ffb80cfb3 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f780fdeb554c8eb1fd91382e05530b : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c71d7c8227644bfcb3ac245d64e1a6e5 : SUCCESS in 1m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06d2d1f6fed54325bbb0ba8db65b932f : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/455cb0cd90c2407dbee7f3275639b653 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68bfe45c38df4afc97dae481dd3f1270 : SUCCESS in 5m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38fe9d577b234cbc8ef2cf4eb6d9408b : SUCCESS in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1b3283b685c7673a1073d79791408a52927bafe","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-15 12:36:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e952c1d67b7845ee5b6d43666024274ea876ec92","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-05-15 15:08:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9053198614c4997866687bc3bc6d4c6f4ca1aa3d","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-27 05:37:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca8ce313478d4e65aa585e4268b105b0173e220c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:37:39.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3c75be44a14d88ef3ea6abf2335448a2acd2f902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:37:47.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"382ca85bfed3db99da103574e5402cc95cdebf80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:22:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865ac1767248462789fe6c48bca11d8e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/322958e71bdd4b0fb3e79e7723fddc1b : SUCCESS in 1h 37m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0610450b37e42de83c0219df2a3bb31 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eaca74c75c04d898cd825dc59146670 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0999be721e7e4ee68d1ae309e2de36ea : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12eefd71b6644b8aa9497118066edabe : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/484b285ed08243b2b9fa4cb707a35b74 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c99dcf407574ae3905301f3e7cec315 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70bbb54e62404326b7d4f0e8e0054e53 : SUCCESS in 2m 54s\n- python-cinderclient-functional-py311 https://zuul.opendev.org/t/openstack/build/c1c06cf09217436d93960a683208547c : SUCCESS in 13m 33s\n- python-cinderclient-functional-py313 https://zuul.opendev.org/t/openstack/build/c5a921197eb64f05a000af8ccafb98b5 : SUCCESS in 30m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bc8f6da8080f4f9384a458435a264fcb : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"773f5d344faa146be24758791283a87dbd30ef59","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-10-02 15:48:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b7924d93398dec76cb06f6fb2fe5f44666f8fa8","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-10-02 15:48:24.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0006737d605a64b30e4ae85e6f37c6ddf04d986f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 15:48:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"15b0286984235e2ea47ff480ce3a9caeb324d25a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:41:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c04998a39f8400f8d4f82c69367cb5d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bc75bf587d94e5a8f6b04627d561536 : SUCCESS in 47m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c915d3b767412b9d1be2dd85d50012 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/278cb06ffcc44115ada8246586181fdc : SUCCESS in 2m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/91c42e30c7ea43c98359b165ac14d44d : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd3e0192e4b46c1bf9ddf774ae9310e : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/779ed5c7d0654cd9b7f97b7fdac0c1dd : SUCCESS in 5m 20s\n- python-cinderclient-functional-py311 https://zuul.opendev.org/t/openstack/build/3c6f893a56b3418ebf2df56935f04e60 : SUCCESS in 15m 01s\n- python-cinderclient-functional-py313 https://zuul.opendev.org/t/openstack/build/c067ab6105274693baa7746c4c16cd91 : SUCCESS in 19m 09s","accounts_in_message":[],"_revision_number":4},{"id":"549568200d5a31d8c016bfc9e3c04c5294eb8741","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:41:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5604f8650ae1c58b2b1901cdd929bf698a16646b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:42:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472085b0f01349e3a6171c7c90936bb2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d949a0cab63d481c84a7eb40e56fc725 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5cc3913f3b27bc0214f9f475d9cf49c1d6d13a92","revisions":{"e8b60d150881f8cb5cd6c6ea136637a335adbfb9":{"kind":"REWORK","_number":1,"created":"2026-04-29 23:27:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/986728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-cinderclient","ref":"refs/changes/28/986728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/1"}}},"commit":{"parents":[{"commit":"3100de0e712f9c17028ad3e09b63a04ea3d3ca15","subject":"Add periodic jobs for cinderclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/3100de0e712f9c17028ad3e09b63a04ea3d3ca15"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-29 23:27:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-29 23:27:01.000000000","tz":0},"subject":"Fix for typo","message":"Fix for typo\n\nChange-Id: Ia7891c1be82d75d023eef2cc3372cac71847101a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/e8b60d150881f8cb5cd6c6ea136637a335adbfb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/e8b60d150881f8cb5cd6c6ea136637a335adbfb9"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"dec32aac4ae849011777abd940aef8b0074a326d":{"kind":"REWORK","_number":2,"created":"2026-04-29 23:27:44.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/986728/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-cinderclient","ref":"refs/changes/28/986728/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/2"}}},"commit":{"parents":[{"commit":"3100de0e712f9c17028ad3e09b63a04ea3d3ca15","subject":"Add periodic jobs for cinderclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/3100de0e712f9c17028ad3e09b63a04ea3d3ca15"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-29 23:27:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-29 23:27:44.000000000","tz":0},"subject":"Fix for typo","message":"Fix for typo\n\nChange-Id: Ia7891c1be82d75d023eef2cc3372cac71847101a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/dec32aac4ae849011777abd940aef8b0074a326d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/dec32aac4ae849011777abd940aef8b0074a326d"}]},"branch":"refs/heads/master"},"98d03c0dc88331920e45a6e362fd71a9a811bb09":{"kind":"REWORK","_number":3,"created":"2026-09-01 19:37:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/986728/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-cinderclient","ref":"refs/changes/28/986728/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/3"}}},"commit":{"parents":[{"commit":"3100de0e712f9c17028ad3e09b63a04ea3d3ca15","subject":"Add periodic jobs for cinderclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/3100de0e712f9c17028ad3e09b63a04ea3d3ca15"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-29 23:27:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:37:39.000000000","tz":0},"subject":"Fix for typo","message":"Fix for typo\n\nChange-Id: Ia7891c1be82d75d023eef2cc3372cac71847101a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/98d03c0dc88331920e45a6e362fd71a9a811bb09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/98d03c0dc88331920e45a6e362fd71a9a811bb09"}]},"branch":"refs/heads/master"},"5cc3913f3b27bc0214f9f475d9cf49c1d6d13a92":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-01 19:37:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/28/986728/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-cinderclient","ref":"refs/changes/28/986728/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-cinderclient refs/changes/28/986728/4"}}},"commit":{"parents":[{"commit":"3100de0e712f9c17028ad3e09b63a04ea3d3ca15","subject":"Add periodic jobs for cinderclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/3100de0e712f9c17028ad3e09b63a04ea3d3ca15"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-29 23:27:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:37:47.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ia7891c1be82d75d023eef2cc3372cac71847101a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/5cc3913f3b27bc0214f9f475d9cf49c1d6d13a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-cinderclient/commit/5cc3913f3b27bc0214f9f475d9cf49c1d6d13a92"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
