)]}'
{"id":"openstack%2Fpython-designateclient~1004224","triplet_id":"openstack%2Fpython-designateclient~master~I903de10e2e52a7a42a76bc986b2b1582f35c4c97","project":"openstack/python-designateclient","branch":"master","hashtags":[],"change_id":"I903de10e2e52a7a42a76bc986b2b1582f35c4c97","subject":"tox: Drop redundant passenv","status":"NEW","created":"2026-09-05 18:09:20.000000000","updated":"2026-09-05 19:26:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cd73503fcf3c135831ecd7244238e2d4d87e51e5","_number":1004224,"virtual_id_number":1004224,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-05 19:26:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 19:26:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fe56d953cf534680d12b7cd714673cf1c278856","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:09:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"133c8d6709462add1ab1752104778cdf83433bac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:13:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cd73503fcf3c135831ecd7244238e2d4d87e51e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 19:26:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edb05d17abb8458183aaa4c01c08973f\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/ae5842890d544c29a0c55d5f394c3f7f : SUCCESS in 40m 47s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b368500ee60241b088af58a8d99c6bd7 : SUCCESS in 39m 30s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/5bbd5d56063e47f19e852571f084abb1 : SUCCESS in 39m 26s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/7bb721191ee24cbe9e63024e3c32338e : SUCCESS in 18m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7cec18e10c0e46c9b684dd7ee88b229a : SUCCESS in 1h 00m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/730da0126d32424694250f0a9ed8b90b : SUCCESS in 1h 07m 57s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/dd0e9b286fc84594a6e2535809bb31cd : SUCCESS in 57m 35s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fd13eb4afeed4635832dc2c07c876b68 : SUCCESS in 52m 45s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/a877a110d1664889b6ae74cb3ba3fd95 : SUCCESS in 36m 21s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/76a14a362a3f4adfa5a5ed5c0d479e56 : SUCCESS in 46m 23s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/41a8a33abcb742b79fb137b87eafaf6e : SUCCESS in 42m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81d3a0460f184347b60132ac000869fe : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44dad73292004cd99661c958e0a6e382 : SUCCESS in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09e908fa9aff4810bd778d74a1109356 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89679fd4e8974125adb71c01a0e9c952 : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a6846182d0004786a370dc839d39b09d : SUCCESS in 9m 09s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/dd658e74a635411eb7095c55ff3aa504 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d21c28214774ed38e5593501ca38c59 : SUCCESS in 2m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/600a109bedec49df9ffec4a890c5c62b : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae38bd5185f066b8f2a5fa998b9ced116ccafa7b","revisions":{"a5f8bca109a97544522d18cc2325583a2cb9c22c":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:09:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/1004224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-designateclient","ref":"refs/changes/24/1004224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/1"}}},"commit":{"parents":[{"commit":"944e373b43ba1474b5c677e163de6de027286298","subject":"Add --status filter option to zone transfer request list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/944e373b43ba1474b5c677e163de6de027286298"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:09:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:09:11.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https: //tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I903de10e2e52a7a42a76bc986b2b1582f35c4c97\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/a5f8bca109a97544522d18cc2325583a2cb9c22c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/a5f8bca109a97544522d18cc2325583a2cb9c22c"}]},"branch":"refs/heads/master"},"ae38bd5185f066b8f2a5fa998b9ced116ccafa7b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-05 18:13:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/1004224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-designateclient","ref":"refs/changes/24/1004224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-designateclient refs/changes/24/1004224/2"}}},"commit":{"parents":[{"commit":"944e373b43ba1474b5c677e163de6de027286298","subject":"Add --status filter option to zone transfer request list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/944e373b43ba1474b5c677e163de6de027286298"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:09:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:13:47.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I903de10e2e52a7a42a76bc986b2b1582f35c4c97\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/ae38bd5185f066b8f2a5fa998b9ced116ccafa7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/ae38bd5185f066b8f2a5fa998b9ced116ccafa7b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
