)]}'
{"id":"openstack%2Fpython-designateclient~996890","triplet_id":"openstack%2Fpython-designateclient~master~Ic17847cc6042486d120a005d57d06adb3ace4964","project":"openstack/python-designateclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-23 13:26:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic17847cc6042486d120a005d57d06adb3ace4964","subject":"tox: Drop redundant override of install_command","status":"MERGED","created":"2026-07-11 17:08:50.000000000","updated":"2026-09-23 13:29:07.000000000","submitted":"2026-09-23 13:26:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996890","meta_rev_id":"dd323fdc50145d64f142ef3bae6159c1477271f8","_number":996890,"virtual_id_number":996890,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 13:25:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-09-23 12:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-09-23 12:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-11 18:26:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 12:00:14.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"c01a4d8f0bfb1d6fb940b599218e14ae9ad84a05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:08:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdd15275b806659955207af23e68b6393699dbc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:09:09.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c8675a9c071fb162bd58122a684a6bba000b5af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 18:26:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/438ecf0f513f4e2594cee40183d29c24\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/675446a122934fb9a5a7e6880d3c25d7 : SUCCESS in 35m 23s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/14305c6da11a4ed7b58979f675247ae4 : SUCCESS in 36m 15s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/839c0ed6de6e4878b010af06e096e4d0 : SUCCESS in 48m 35s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/80bb1eda65fc4136805dcfff0d850d2d : SUCCESS in 17m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e952e702c72b4e2d922c73b6911a7c9f : SUCCESS in 1h 07m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/83636828a4654686b0041009c8228461 : SUCCESS in 1h 09m 13s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cbf0af714618430a9fe09f6619cfe017 : SUCCESS in 52m 37s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2b8f4d77da3f4e05a87ce7839fe9a2ed : SUCCESS in 1h 11m 19s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/43cf07ed97cf489393e4a16b6dfcb609 : SUCCESS in 43m 01s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/52b3f1ce41074290a283133c20de9098 : SUCCESS in 35m 28s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/78924004f7b045c199ba7ce84e5fac6b : SUCCESS in 39m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/226c90915214477eba75356a346ee669 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/173ae601a6dc45948c16b6a7938d076a : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcf4b6ba28c042fcbf5090281054f271 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/242a91d6e52b45829a73523c857b13f3 : SUCCESS in 2m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d3f7ac82af04cc48a56f9399921a6fd : SUCCESS in 5m 09s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/bd55b180d5ff419989469d4d5e02beb3 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/530305c3d1b44deca0f79982f1f9f342 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aec4ca0a8e234707972f930fbfe199f6 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4d43c2a447b0505a143ca1b161070e693f1fd002","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-23 12:00:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bb8ef6b41ce74be6372d86d88c531b76d40dbe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 12:01:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eed8c70a14f2200a50167b07ad49dec07a06f662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:25:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7574579649a1406aa781f4b5aa4e5e0d\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f01bb5617cef449f903682f08c5ff653 : SUCCESS in 48m 27s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6897d5054ee14f4286ccd737aa51ed12 : SUCCESS in 46m 18s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/beedc33200824476835b0839b5071a13 : SUCCESS in 42m 23s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/4af21a636ccc40b5a9c796f3672aa18d : SUCCESS in 35m 04s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/e6221b4fc25c472cbbec35b7d80e8efa : SUCCESS in 33m 08s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/3677c7e680864d25a6657cbd5b6a4375 : SUCCESS in 42m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cc6e53839d4c778c45b278205813cc : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e506a5e9d0a46bc8f6f20e5bbd82b46 : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59b2afab6b7843aa9d85f5418ce4b215 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4aac32b1f24a108128d6a3df77b79e : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ed3a389870a47118e32abb8b6ff059c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8a1bb2834030959bd1a6877cf7ee9531f2e48e79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:26:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dd323fdc50145d64f142ef3bae6159c1477271f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:29:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93f569f4cc0e4689a6bfb03a2ad44747\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8707dc290f04762be77ff9220913f57 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5aa14b8a4704f0f845a503008c76d86 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"476c931e33aff56ffff3428b9a947c1d94b35876","revisions":{"3747375d29e4df26ec16a457a760bc3f7aa33e77":{"kind":"REWORK","_number":1,"created":"2026-07-11 17:08:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/996890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-designateclient","ref":"refs/changes/90/996890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/1"}}},"commit":{"parents":[{"commit":"0dd5ca4489bd7d2db3e3db34758ee3fc25f07da4","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/0dd5ca4489bd7d2db3e3db34758ee3fc25f07da4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:07:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:07:47.000000000","tz":540},"subject":"tox: Drop redundant override of install_command","message":"tox: Drop redundant override of install_command\n\nIt is almost equivalent to the default value:\n\n  python -I -m pip install {opts} {packages}\n\nAlso remove redundant VIRTUAL_ENV environment and fix inconsistent\nindentations.\n\nChange-Id: Ic17847cc6042486d120a005d57d06adb3ace4964\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/3747375d29e4df26ec16a457a760bc3f7aa33e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/3747375d29e4df26ec16a457a760bc3f7aa33e77"}]},"branch":"refs/heads/master"},"476c931e33aff56ffff3428b9a947c1d94b35876":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-11 17:09:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/996890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-designateclient","ref":"refs/changes/90/996890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-designateclient refs/changes/90/996890/2"}}},"commit":{"parents":[{"commit":"0dd5ca4489bd7d2db3e3db34758ee3fc25f07da4","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/0dd5ca4489bd7d2db3e3db34758ee3fc25f07da4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:07:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:08:55.000000000","tz":540},"subject":"tox: Drop redundant override of install_command","message":"tox: Drop redundant override of install_command\n\nIt is effectively equivalent to the default value:\n\n  python -I -m pip install {opts} {packages}\n\nAlso remove redundant VIRTUAL_ENV environment and fix inconsistent\nindentations.\n\nChange-Id: Ic17847cc6042486d120a005d57d06adb3ace4964\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/476c931e33aff56ffff3428b9a947c1d94b35876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/476c931e33aff56ffff3428b9a947c1d94b35876"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
