)]}'
{"id":"openstack%2Fpython-designateclient~997192","triplet_id":"openstack%2Fpython-designateclient~master~Ifd951076718fc77dbc3ef7742f25c887be8379cb","project":"openstack/python-designateclient","branch":"master","topic":"add-missing-f-string-to-nouniquematch-exception","attention_set":{},"removed_from_attention_set":{"30314":{"account":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"last_update":"2026-09-23 13:25:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd951076718fc77dbc3ef7742f25c887be8379cb","subject":"Add missing f-string to NoUniqueMatch exception","status":"MERGED","created":"2026-07-14 10:30:15.000000000","updated":"2026-09-23 13:28:04.000000000","submitted":"2026-09-23 13:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997192-add-missing-f-string-to-nouniquematch-exception","meta_rev_id":"4136f9a795a5c9a3bac47f1724d445fadda932fe","_number":997192,"virtual_id_number":997192,"owner":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 13:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-09-23 11:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2026-07-14 11:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-09-23 11:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-14 11:07:05.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"REVIEWER"},{"updated":"2026-07-14 11:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 11:59:04.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"77e0739a28930cc938ef1b00ed94a46406c3cb0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2026-07-14 10:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e62c8f990488391c4d4f16bc9cc6a8d3a6d16a","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2026-07-14 11:07:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1159dec1981ecc31daef531a5c46c31b4ecbf079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 11:50:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54dd278c370477bbc891a183c382205\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/fabef684f8194b12ab3620356b3fd348 : SUCCESS in 36m 25s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e56f83ffbe184ae0aa83732e5eac56ee : SUCCESS in 31m 26s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/6cd1cad1e65b46e99a16823b47214125 : SUCCESS in 46m 30s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/344c13e22c974ab2bf70cb4cb0a3f2a6 : SUCCESS in 31m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b9fddda7e8844aa0a4abd5ab2aca7025 : SUCCESS in 1h 09m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78b419d7a3f14a3da4d83b338e0e5fbb : SUCCESS in 1h 13m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/82cd85d300634a6b9994cb1dfede787f : SUCCESS in 1h 11m 32s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/276033b9b1074a7f8b3d1d649c81102f : SUCCESS in 1h 06m 35s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/121b9e4dafe34a83ad5b082441413d94 : SUCCESS in 39m 49s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/04e5089c7b954dc6936535d55c9d914e : SUCCESS in 42m 50s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/510c86aea5e642ddb0352d6fb0cb0803 : SUCCESS in 46m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2119c52f6b484e24bc7b9517ed59e57a : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fed68854df4a76b162ab4a47080960 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/061a44e7ccdb40398d56df1660f76eb3 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1d4bac8aa0d4c59b2286db7223786f5 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0354d53386ca43b4875fa870856b2778 : SUCCESS in 5m 33s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/844395a0d03043b789f20ee0219c69ae : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8360102b911434495d16bcb5bfc71c5 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/828e207c69ec475d90a66ca60293f527 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4e125b49a25648d282a15344ef0bc678c9a02d3d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-23 11:59:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a4ae336a23c60ed3ce81946aabfbde7234f42a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 11:59:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba367bcb0b12438999071adae8a44065f09b39a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:25:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91109edb952f4d609611059d09331b4c\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/2dffd71802b242fe911ac7990ecba347 : SUCCESS in 49m 59s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ff9b6ce9c17a47658c905fefa3968a97 : SUCCESS in 39m 30s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bd166358e6674e84b3d3c9fc5b9f7c47 : SUCCESS in 45m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d75953f136384036af170726d27507de : SUCCESS in 33m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4952072a1b53474481bc2c2c977186ca : SUCCESS in 1h 13m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0c7fd54f43f344178fc71b33730b249f : SUCCESS in 52m 06s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/11342da2cc25480c9bd08ffc63f90a1d : SUCCESS in 1h 04m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/da7a999abbdc4c95beec4772768bea07 : SUCCESS in 57m 01s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/f20c8ac10add4deda8d94e80c02e9301 : SUCCESS in 45m 40s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/6fcc4107bece473983163e3e4975a3c8 : SUCCESS in 46m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d02df9ccdf694d0fb44eafe457a7cef2 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb8ceb67d4ee45eea6ab14158fa4d77f : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/01b21a21bffc4a4b9535ef9df89ccd57 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b7d41ad9e54f3081bfadf84b120409 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c72bb017c254ff9a3477c22bbdad106 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"13b84eb6af3a870e26af1130e3298a730032eda3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:25:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4136f9a795a5c9a3bac47f1724d445fadda932fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:28:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d8abfd5a66c4891b9eebc514db1fd03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb3c1f2fe8743d1ac29e1e8dbbdbaf2 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f3383654fedab0411e3f0bc7c53031a05316b92","revisions":{"7f3383654fedab0411e3f0bc7c53031a05316b92":{"kind":"REWORK","_number":1,"created":"2026-07-14 10:30:15.000000000","uploader":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"ref":"refs/changes/92/997192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-designateclient","ref":"refs/changes/92/997192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/92/997192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/92/997192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-designateclient refs/changes/92/997192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-designateclient refs/changes/92/997192/1"}}},"commit":{"parents":[{"commit":"944e373b43ba1474b5c677e163de6de027286298","subject":"Add --status filter option to zone transfer request list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/944e373b43ba1474b5c677e163de6de027286298"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2026-07-14 10:23:30.000000000","tz":120},"committer":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2026-07-14 10:29:57.000000000","tz":120},"subject":"Add missing f-string to NoUniqueMatch exception","message":"Add missing f-string to NoUniqueMatch exception\n\nWhen NoUniqueMatch is raised for multiple results being found the\nexception text was not properly interpolated due to a missing `f` in\nfront of the string.\n\nChange-Id: Ifd951076718fc77dbc3ef7742f25c887be8379cb\nSigned-off-by: Sebastian Lohff \u003csebastian.lohff@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/7f3383654fedab0411e3f0bc7c53031a05316b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-designateclient/commit/7f3383654fedab0411e3f0bc7c53031a05316b92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
