)]}'
{"id":"openstack%2Fpython-glanceclient~999283","triplet_id":"openstack%2Fpython-glanceclient~master~Ic3a6ea48f7f083a7944867d9d4ce3cc14fd6f529","project":"openstack/python-glanceclient","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-26 00:59:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3a6ea48f7f083a7944867d9d4ce3cc14fd6f529","subject":"Bump hacking","status":"MERGED","created":"2026-07-30 13:30:41.000000000","updated":"2026-08-26 01:00:08.000000000","submitted":"2026-08-26 00:59:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999283","meta_rev_id":"9fca196bd9f130db52a605bd15b6ebdaa6c4b6e4","_number":999283,"virtual_id_number":999283,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 00:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-08-03 14:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-08-25 15:30:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-30 15:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-03 14:11:03.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-08-25 22:35:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"3215242bfdf69d09b4dfd086c56185b1d0625033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-30 13:30:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f586841c219af806132b6119c25768d79e1ecddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 15:02:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c524cf7dff6e4287b12fd6f72a71f1d8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/56a2016e56cb438fb43594b96e0113a7 : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12a770c77ed24f41a4dfce0d354ab8a4 : SUCCESS in 1h 23m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ba5bb76202345559466d76c4ef44f5e : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbed0eeb8f214b9d98c87e2f7d2bcab9 : FAILURE in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0ef6b1d64a44d9d8d82aa14dcb1bbcb : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57d43a027e88486ea73c2a1293d56647 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03d1cfc3dfd1410eafeccfd1afdfd7b1 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57caead7cd6944e0bd392a3a82f5dd37 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9037ddec53f2443a88eaebd5d09809c0 : SUCCESS in 4m 35s\n\nWarning:\n  Comments left for invalid file glanceclient/tests/unit/test_shell.py\n  Comments left for invalid file glanceclient/tests/unit/v2/test_shell_v2.py","accounts_in_message":[],"_revision_number":1},{"id":"27d2a971bc4edff9b2dd49afd995df7e36223ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-30 16:41:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1bf22b4aae37dd6d4932ce2d350658ef7d3e5a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 18:34:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5adfe2bce20c46efaf878460ef60c30f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b7a90235ee849938932f47a5ceabfd4 : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ec26885612749118929d71ae45517eb : SUCCESS in 1h 43m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2664eaf88b4c4af79427df4e57ec6187 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5040ef72f141128e25457af6711d7e : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd37abe99dbd472f81b9aa99924a6650 : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b435a6873a040b0b67a9d5acbec56da : SUCCESS in 6m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c380d12941a643be90cfb6aa841e64ec : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25938dcf7ff48e78e671da39a4bcb30 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b39ee7ac3d4771a6e308f3933f9b80 : SUCCESS in 3m 07s\n- glanceclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3231b80a4fc49a49fa8031c776bc09b : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a7242c9f522ea89885cc0babc8240ca9b830ceae","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-03 14:11:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d617cd80a970a6e81045459ff877cfedb057d0d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-25 15:30:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"466e39c859ca120e81f0c458b2f56a6e466ce6f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 15:31:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aecdff81a19d8ffa2a711539695ceaee9b2b0f68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 16:31:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16028951249148b4b80ef6035cfeb302\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/30882b7b3d66479484a054ceaf0d9763 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31498e66fb5743b9bf4e1b2699eddf7f : FAILURE in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d3cad3505c49f79e7da0417194d040 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff8c44940a254f468e96bb9aedd89d2b : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd7f42d7fc4c4ebd9c0ee580e5237144 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124414867a4e49d195d8df087a08ac53 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82fa9b626dee4020b74a95dc2379a739 : SUCCESS in 4m 08s\n- glanceclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e37ad3888d145748f54b0468da7c8e8 : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c3811f0d946964526eb51ba5d6a6a9124617426e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-25 16:58:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b49dafe62cf41f4584688239d8e8b6c7614546ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:14:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcef2dd66b8b4043b791be5af36ff9e6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f423856c29d74bb395328c799d169789 : FAILURE in 5m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d3a34aa87b14c36a7c0c07ce3cd11bb : SUCCESS in 1h 10m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6940e01953164f39a5c65b702e18c0c8 : RETRY_LIMIT in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a1fc1124a24b45b47fc8054f819507 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5ad2e0718a14dcfb0277b0da127a244 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a4bbe0c899a4dc1ab6c321cf133844c : SUCCESS in 6m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2dd0b971b07748a58f487c35b28048e9 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986dc18470d445a394b15cee43315223 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86b4b9e5faac409f9ecc73489f21ef24 : RETRY_LIMIT in 5m 19s\n- glanceclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6e678ba520c45ffb95aa207c33d48c9 : POST_FAILURE in 8m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4ba9141c4f8c3400adc623e991b36e8705949b35","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 22:35:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"837ec9bda40e086d423f519d89a6ea2b71168631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 23:58:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbeb78d7981e47f88f09ab3400ef4c84\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec7112954ce34c0fbf55bb1fb915f1a3 : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dd179076b264cd0933bbb6dfb280896 : SUCCESS in 1h 20m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06392fb6ea37405caab6b2706cfaf3c8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/295ee2e8a52f4423957d27c24a758a7a : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/224255f50bf541eebada1945a11bb6bf : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06a609f2c8704372ad73b39096dde730 : SUCCESS in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd78100a72f94affbad02788b45ab5e5 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74619b16e9b14f53b76931be655b160e : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a59f08ecb4084841b89f9b3c7d15bda5 : SUCCESS in 5m 10s\n- glanceclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b1d31191d674d328b47129a0d14073e : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3df22256b5520882bc0431334bba343cd4b96874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 23:58:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a57da0698c9d26a74349b9e10da0b5aa251a475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:59:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac91e19550544aeabd9e0464f6f5c04\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b34bc79fb22d42b681a1a52bb53878f5 : SUCCESS in 1m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/549f29aa33a5469985a7cda16a9513ec : SUCCESS in 1h 00m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d086542ec02418c932df21358876d2d : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1656444ad69c42a68a814a18a6257430 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b0c545dbc2145a59def26bab07f6b5e : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916c250bbec34254bc4319f025b94747 : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31e7e757902c462cb0615815f6a30174 : SUCCESS in 2m 10s\n- glanceclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfb18ab61c7b4dc6900cc82b3e73431b : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d5f2abab28cd966e53cb676e6b2b502271e345ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:59:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9fca196bd9f130db52a605bd15b6ebdaa6c4b6e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 01:00:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8ec412e9ff449294e4c8aaa9f40b41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75951a2396c4775a57bb84ce9a2d6cb : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31cacc9d976841dabe4da6362230eb0b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8bbd6eedc759ddc330107cf784f36adfe7e63f8","revisions":{"bbbb2b0adffc03710c417378cbb2f68909a27a10":{"kind":"REWORK","_number":1,"created":"2026-07-30 13:30:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/999283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-glanceclient","ref":"refs/changes/83/999283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/1"}}},"commit":{"parents":[{"commit":"50595eea02a2727b45f870f8fe8304c207c0d049","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-glanceclient/commit/50595eea02a2727b45f870f8fe8304c207c0d049"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 13:30:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 13:30:31.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nThis is required to run pep8 check in Python 3.14, which may be\nrequired in the next cycle.\n\nChange-Id: Ic3a6ea48f7f083a7944867d9d4ce3cc14fd6f529\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-glanceclient/commit/bbbb2b0adffc03710c417378cbb2f68909a27a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-glanceclient/commit/bbbb2b0adffc03710c417378cbb2f68909a27a10"}]},"branch":"refs/heads/master"},"e8bbd6eedc759ddc330107cf784f36adfe7e63f8":{"kind":"REWORK","_number":2,"created":"2026-07-30 16:41:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/999283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-glanceclient","ref":"refs/changes/83/999283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-glanceclient refs/changes/83/999283/2"}}},"commit":{"parents":[{"commit":"1e7beb36199a3058094df1bdf7d444100adb412a","subject":"Merge \"Add pyproject.toml to support pip 23.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-glanceclient/commit/1e7beb36199a3058094df1bdf7d444100adb412a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 13:30:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 16:41:29.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nThis is required to run pep8 check in Python 3.14, which may be\nrequired in the next cycle.\n\nChange-Id: Ic3a6ea48f7f083a7944867d9d4ce3cc14fd6f529\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-glanceclient/commit/e8bbd6eedc759ddc330107cf784f36adfe7e63f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-glanceclient/commit/e8bbd6eedc759ddc330107cf784f36adfe7e63f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
