)]}'
{"id":"openstack%2Fpython-ironicclient~697271","triplet_id":"openstack%2Fpython-ironicclient~stable%2Frocky~Ib62ca3ee4626084e5e9b90e93e4fa97938023457","project":"openstack/python-ironicclient","branch":"stable/rocky","topic":"story/2006393-stable/train-stable/stein-stable/rocky","hashtags":[],"change_id":"Ib62ca3ee4626084e5e9b90e93e4fa97938023457","subject":"Fix AttributeError in negotiate_version","status":"ABANDONED","created":"2019-12-04 11:26:10.000000000","updated":"2020-03-12 16:26:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"966803becbd417372f1f4d5c288c70063eeaf4b3","_number":697271,"virtual_id_number":697271,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-19 16:36:44.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"date":"2019-12-16 15:50:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-04 12:25:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-05 11:01:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-06 06:00:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 12:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 11:01:51.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-12-06 06:00:39.000000000","updated_by":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"reviewer":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"state":"REVIEWER"},{"updated":"2019-12-19 16:36:44.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"b2bc7f1739bf51752be831d223d0ebd36ee183dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-04 11:26:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"b3777d44733ffdc8ca18bfb2ae38c76cdf895fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 12:25:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/12d756e5a96e4bc3849be457d8656e01 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3176fdc536eb443a893260c02c36edcb : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/372cfe20bbad4e929974274bf88b643c : SUCCESS in 3m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fedecb148ec34135a383973bc52559c0 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6de00714e404c96838bdb3f581c6e86 : SUCCESS in 3m 46s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e683f3e348d84f3aa87c8364979206a0 : FAILURE in 5m 35s (non-voting)\n- ironicclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/7370975443a34d2b98beb4c95e7259cf : FAILURE in 21m 31s\n- ironicclient-tempest-dsvm-src https://zuul.opendev.org/t/openstack/build/8e14450bafb44042b28c15fd7688c07c : SUCCESS in 50m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c51f280c4d3245f899aa2367652a5341 : SUCCESS in 5m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/023bd565f4d54264899d6dd6f6dce814 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c38e6065197e54fd64a55138742b614dd1a38cca","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-05 11:01:51.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10a082a0ba2311d875641300b19c300f4a481180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 11:50:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/031eb43bf34f485a8dbff5f730ebfd24 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31c61394b0704567a7ee7fce47f84320 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fc63cb9dfb144e898b80f59ab157c77 : SUCCESS in 3m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fedbc52172524f9d840096372cc1203c : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b410c038648404db6f0c578b1e4ebc9 : SUCCESS in 3m 30s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/99c87a7a67424c319813eb3a51204409 : FAILURE in 4m 17s (non-voting)\n- ironicclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5515ea19f3d4a2294d88238d780677c : FAILURE in 22m 03s\n- ironicclient-tempest-dsvm-src https://zuul.opendev.org/t/openstack/build/28d11782fa5c4df89590daae0a2f4e2f : SUCCESS in 46m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6af044bc8b84c7897916738a2290b4c : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33f7cde235834a9db62d50772d56b5c7 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7555ede65efe05e782d44f11744487d3e84e72d5","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-12-06 06:00:39.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e720eaab33664af598e984794ec987ab65507d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 07:10:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1f064a15e5f94f0ca4338de41a05954c : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b117e1f1a3b6455aa004b50c2ce4e92b : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a39f98c49e4c4491952c45eab46278a4 : SUCCESS in 5m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/73666aec90ed4edf9fb262343830fc6f : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c22fa77732844d5785817fa42ed7c24f : SUCCESS in 3m 46s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c1696708925c47e29ac655a3740cc37c : FAILURE in 7m 10s (non-voting)\n- ironicclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4f837eab104414c9ea9d1c3d244d781 : FAILURE in 26m 12s\n- ironicclient-tempest-dsvm-src https://zuul.opendev.org/t/openstack/build/ad7cbb8d5e1f4cababab39d3917148f6 : SUCCESS in 57m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/925ccfc9fcfa4381adacad5b1248687b : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57dff7392e15490686f60c8f63e09338 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8ae2472796a29a4545f47b3b2fafbb0eb13b20a8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 15:33:52.000000000","message":"Patch Set 1:\n\nSeems to be failing in openstackclient-check-plugins due to sahara:\n\nSome commands failed to load...\n{\u0027dataprocessing_data_source_delete\u0027: [\u0027saharaclient.osc.v1.data_sources","accounts_in_message":[],"_revision_number":1},{"id":"868f75d1773ca268c61b2e8b854df143b0855eb1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 15:50:24.000000000","message":"Patch Set 1:\n\nA bunch of unrelated functional test failures:\n\nhttps://storage.bhs1.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_d4f/697271/1/check/ironicclient-dsvm-functional/d4f837e/logs/devstack-gate-post_test_hook.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"f743561efd674f9027830790d831c32bf718582f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-17 08:11:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3747450a025fbb7de45ed5725034479013db641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 09:08:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/302d3f77c8254f699ce64a64df6eae2c : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8baead58fb490fb91bde374ad0e479 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c048844afa914f759ec92f48e8c02570 : SUCCESS in 5m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/55f3809f9beb4c20ae0f9ccb4b924c7c : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c420e12fe1474fed8c4f03ecba986a1c : SUCCESS in 5m 52s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/92a0aeb4fc0d477c972a50b39db472ad : FAILURE in 6m 50s (non-voting)\n- ironicclient-dsvm-functional https://zuul.opendev.org/t/openstack/build/feff8acea89e4085a6277bed71b55603 : FAILURE in 21m 54s\n- ironicclient-tempest-dsvm-src https://zuul.opendev.org/t/openstack/build/3a4c84f3d1144d9cb66f113faa4ddb49 : SUCCESS in 50m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/607e5af0e9944a739bd260b9a2fab2dc : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed1c65491804171b69c60bae2201299 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"86c8ed2831cb293b89983947936f01b50fd77fa1","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-12-19 16:36:44.000000000","message":"Patch Set 1:\n\nsubmitted https://review.opendev.org/700022  to test CI.","accounts_in_message":[],"_revision_number":1},{"id":"720cd3999a8cba86317932de5fcf000f69e8f570","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 16:26:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"114962db580c93536ebef8547c6654aa9454a368","revisions":{"114962db580c93536ebef8547c6654aa9454a368":{"kind":"REWORK","_number":1,"created":"2019-12-04 11:26:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/697271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-ironicclient","ref":"refs/changes/71/697271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/71/697271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/71/697271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/71/697271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-ironicclient refs/changes/71/697271/1"}}},"commit":{"parents":[{"commit":"877875a62fee39d8d287fae144c907e4a4fd146d","subject":"fix session cert arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/877875a62fee39d8d287fae144c907e4a4fd146d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-12 15:19:37.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-04 11:26:10.000000000","tz":0},"subject":"Fix AttributeError in negotiate_version","message":"Fix AttributeError in negotiate_version\n\nSeen in kolla-ansible CI, if ironic inspector starts up before ironic\nAPI is properly up, we see the following error:\n\nThe PXE filter DnsmasqFilter, state\u003dinitialized encountered an\nexception: StrictVersion instance has no attribute \u0027version\u0027; resetting\nthe filter: AttributeError: StrictVersion instance has no attribute\n\u0027version\u0027\n\nExample: http://paste.openstack.org/show/756342/\n\nThe usual cause of this error is when StrictVersion is initialised with\na version of None. This suggests to me that max_ver is None.\n\nThis leads to an exception being raised that circumvents the client\u0027s\nretry mechanisms.\n\nIn the particular case of kolla-ansible CI, Ironic API is behind a load\nbalancer, and all backends are down resulting in a 503. The lack of a\nretry caused ironic inspector to fail on startup.\n\nThis patch catches the case where we get a 4xx or 5xx return code when\nchecking the version, and raises an error that works with the client\u0027s\nretry mechanisms.\n\nChange-Id: Ib62ca3ee4626084e5e9b90e93e4fa97938023457\nStory: 2006393\nTask: 36266\n(cherry picked from commit 0e9bf337f1f6aa41f67df2966dacf9e8e897b7da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/114962db580c93536ebef8547c6654aa9454a368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/114962db580c93536ebef8547c6654aa9454a368"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
