)]}'
{"id":"openstack%2Fpython-ironicclient~703982","triplet_id":"openstack%2Fpython-ironicclient~master~I5fe379c4ff439b3a083ae819ce5b4bdbddd22b3a","project":"openstack/python-ironicclient","branch":"master","topic":"story_2005425","hashtags":[],"change_id":"I5fe379c4ff439b3a083ae819ce5b4bdbddd22b3a","subject":"Add support for retired{_reason} fields.","status":"MERGED","created":"2020-01-23 12:22:35.000000000","updated":"2020-02-11 10:45:36.000000000","submitted":"2020-02-07 15:17:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703982-1581088649376-85323f19","meta_rev_id":"b790a9aaa8bea05985a4fd444ce9ee55b4c0244a","_number":703982,"virtual_id_number":703982,"owner":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 15:17:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-02-06 23:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":2,"date":"2020-02-07 13:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":1,"date":"2020-02-07 13:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 14:52:42.000000000","updated_by":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"reviewer":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"state":"REVIEWER"},{"updated":"2020-02-05 07:59:08.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-02-06 23:27:51.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-02-07 13:25:03.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-02-07 15:17:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ebfbab2224b55740ed09530f70c4c8d78cb22b1","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-01-23 12:22:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06fe8b59496d420638baa5e08bf5855aa5099f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 13:32:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/166ffea8b86c46ee92c94b497ecbb480 : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c6ceac09b1e41dea698cec1914e2371 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0ca9c79b134fd88631c903bb9884ff : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a191e1f893f4aa0877439da9aacb4e1 : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4378c7aba08f4729b7851c8ebba07d0c : SUCCESS in 4m 26s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/82c7e0e7e8eb426f9180ffa5b259c5f0 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec06dd4300d442cb4ab2da8cced1cc7 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef8483ddb6f34eb398de6f00c88fd6c4 : FAILURE in 3m 24s\n- ironicclient-functional https://zuul.opendev.org/t/openstack/build/138d494af4bd482eba6baae70399c049 : SUCCESS in 26m 55s\n- ironicclient-tempest https://zuul.opendev.org/t/openstack/build/ae55c3eaf1064a7b9e24e14726ea20cd : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9a67a798c3c7ceee41dd13a734b99c1786ba3be4","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-01-23 14:12:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01bfef99ebbbf9b49f80fa1d2422de2e3ba5be3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:07:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a19e47a574a4fc2a2dd7f83405189cf : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cfc56cb33674841885e9852315b725f : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ecfe467a184545866b0b4cba82e3e5 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e136cacba9b74dbfaccc65feaea4f3c3 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/531a168a3bd84504b5afee4d87ab748f : SUCCESS in 5m 49s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/8a0a64e13d6548b5a1a6d2e0e07f0752 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abbcd310d1ed4cea89d923b8cf20d2f5 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e650d24b536d47428db5efd21d3842e3 : SUCCESS in 3m 47s\n- ironicclient-functional https://zuul.opendev.org/t/openstack/build/e41c2257f6c54758a6b491da63c8f997 : SUCCESS in 26m 42s\n- ironicclient-tempest https://zuul.opendev.org/t/openstack/build/f74361a9f24b4466bdbfaaf7d3a2c482 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":2},{"id":"2facd1a0ebfd0a787e47d86d1fc44d45c1b27da2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 14:08:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49bff6c118fd9e1a9c5695035ddde3651fea7f62","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-01-28 09:17:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"639bfedbc5e92227d21bd8cf6e5c2a3cd74a782b","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-01-28 09:21:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"05a31aa9535ffa2c5a174711496d89db1bcc2961","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-01-28 09:34:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fb0f0d2e4c1f03a153b7d742c7b479caeec37a7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 09:50:06.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"372dd7563857c0d70f203f1342591b102b58fc86","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-01-28 13:45:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b09ab7a5850028899315c955dcc6357a35275062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 14:34:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0d35c2094a8406a937b997578f142b7 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5107dcdfdec417cb8294afa1433d736 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176eb474a85947faa085a11171469b7f : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742807f14839492995550ff0b818e2dd : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/046515e0010e417bbd4721e8b9fbbad3 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57774815289e4960b04a8e8ee8d6dad9 : SUCCESS in 6m 26s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0245369e4df14e50bf188ed51b471b9c : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcfbb6b130af4612bd6edfd2685a4e93 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37545b2354944f0e8abd191c1ccd4f57 : SUCCESS in 4m 12s\n- ironicclient-functional https://zuul.opendev.org/t/openstack/build/3ed3ea319861483c8a189b5b49966df6 : SUCCESS in 33m 51s\n- ironicclient-tempest https://zuul.opendev.org/t/openstack/build/d451bbc6ef2744dca63ae331c871738d : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":4},{"id":"f9972bcd18c9bfbabccc84ca88702827a88ebf52","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-02-05 07:59:08.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1cf633e517ec269643de4ee9be23d6f4bdf91339","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-02-06 15:10:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"14a2dadfb9c5c1175dfbc2c9bff07378780b352f","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-02-06 15:10:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4aa9e498c4125daab4c5c80d83730068186cd86f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 16:00:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b891fa3798b4b428bbcf63c88cc2769 : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f583750dd1ee458996949c484ac7e927 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5964951c24444243809ab0e19e746c56 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5b96b5c528043f0968a5bbc0d2bc0da : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ebf5c9681024286afa5fe0d9f6252ca : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2a66ec5d0e249dbb460c8d50748fe17 : SUCCESS in 6m 56s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/de33a51296a845af94506fd576e0cdc5 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448dc7ba167d4800b7a7cc22ca693a56 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5ef832f70f646ef8c2547a9a66aa9e3 : SUCCESS in 4m 14s\n- ironicclient-functional https://zuul.opendev.org/t/openstack/build/6913fdabf1844ae5a004bb44ccd66c27 : SUCCESS in 20m 18s\n- ironicclient-tempest https://zuul.opendev.org/t/openstack/build/d086ba4696a644c39b613c52f347d9b8 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":5},{"id":"10815f0b7dca46ef5c493fe9e258d7d134987d9d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-02-06 23:27:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6fd53138d2f1589e9257c35c7e94aa8e9cb9b712","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-07 13:25:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1106994353a59be4bbf44c5359ae89b30ab65558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 13:25:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"266c8a6b7de083ce9cb50c2ee4a19d8516af2243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 15:17:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"7456469053e19424ae01882f5fa8433cce0f85e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 15:17:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/441c86d396b6456cb16327bd552379bc : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd85114c323476aa97971e312cb9e5f : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce87afbe6438436bb7734ab6002103b5 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f626e443f2474abcb1d9082e1f7a331e : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd11bfc5c4044f08c3081128c4aea42 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba4747f0259d4e228f2398a8dffe58c4 : SUCCESS in 5m 16s\n- ironicclient-functional https://zuul.opendev.org/t/openstack/build/a84259db5f8146deaf365e54a0390d3f : SUCCESS in 20m 12s\n- ironicclient-tempest https://zuul.opendev.org/t/openstack/build/5f3d9084e849497aae8d15dcaa89ae94 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":5},{"id":"2b8fa2c902f1e8ef847ddddb1ee89efe1bdd5b90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 15:19:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515b914c61774274acc75ec84433ba38 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41681d65cc54410bb6b9b6571f7b2f7a : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":5},{"id":"6651f5752be839df7b4648c11044dd1c15c8e9fc","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2020-02-10 15:16:08.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks Arne! Sorry, I missed the boat wrt the feature, but have a few comments anyway :)","accounts_in_message":[],"_revision_number":5},{"id":"ef06148ee0d6828ae533b56546b0445e7b750951","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-02-10 15:25:22.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8251866dfb84b196ca84bf5f0513c642625a2311","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-11 10:45:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"81eebfc404396345055b50c124962c4069048481","revisions":{"0a289cd2429cc9573d15b5b6e9a4a615218433c8":{"kind":"REWORK","_number":1,"created":"2020-01-23 12:22:35.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/82/703982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-ironicclient","ref":"refs/changes/82/703982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/1"}}},"commit":{"parents":[{"commit":"7d5804984170b2aab7b12bf407094fac965729a8","subject":"Merge \"Enforce running tox with correct python version based on env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/7d5804984170b2aab7b12bf407094fac965729a8"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-17 12:51:37.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-23 12:22:13.000000000","tz":60},"subject":"Add support for retired{_reason} fields.","message":"Add support for retired{_reason} fields.\n\nThis change adds support to set and unset the \u0027retired\u0027\nand \u0027retired_reason\u0027 fields. It also extends the \u0027list\u0027\ncommand with an additional \u0027--retired\u0027 parameter to\nlist only nodes which have the retired property.\n\nStory: #2005425\nTask: #38143\nDepends-On: https://review.opendev.org/703981\nChange-Id: I5fe379c4ff439b3a083ae819ce5b4bdbddd22b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/0a289cd2429cc9573d15b5b6e9a4a615218433c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/0a289cd2429cc9573d15b5b6e9a4a615218433c8"}]},"branch":"refs/heads/master"},"477e05ff1767472a87024feb4e2cff97c38c31bb":{"kind":"REWORK","_number":2,"created":"2020-01-23 14:12:27.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/82/703982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-ironicclient","ref":"refs/changes/82/703982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/2"}}},"commit":{"parents":[{"commit":"7d5804984170b2aab7b12bf407094fac965729a8","subject":"Merge \"Enforce running tox with correct python version based on env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/7d5804984170b2aab7b12bf407094fac965729a8"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-17 12:51:37.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-23 14:08:54.000000000","tz":60},"subject":"Add support for retired{_reason} fields.","message":"Add support for retired{_reason} fields.\n\nThis change adds support to set and unset the \u0027retired\u0027\nand \u0027retired_reason\u0027 fields. It also extends the \u0027list\u0027\ncommand with an additional \u0027--retired\u0027 parameter to\nlist only nodes which have the retired property.\n\nStory: #2005425\nTask: #38143\nDepends-On: https://review.opendev.org/703981\nChange-Id: I5fe379c4ff439b3a083ae819ce5b4bdbddd22b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/477e05ff1767472a87024feb4e2cff97c38c31bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/477e05ff1767472a87024feb4e2cff97c38c31bb"}]},"branch":"refs/heads/master"},"c5a7ec60c22c995893aa11bd05fd9609a85d9217":{"kind":"REWORK","_number":3,"created":"2020-01-28 09:21:36.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/82/703982/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-ironicclient","ref":"refs/changes/82/703982/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/3"}}},"commit":{"parents":[{"commit":"a64370701e8acff6aa28da87ccebd72664a48d34","subject":"Add allocation owner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/a64370701e8acff6aa28da87ccebd72664a48d34"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-17 12:51:37.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-28 09:20:39.000000000","tz":60},"subject":"Add support for retired{_reason} fields.","message":"Add support for retired{_reason} fields.\n\nThis change adds support to set and unset the \u0027retired\u0027\nand \u0027retired_reason\u0027 fields. It also extends the \u0027list\u0027\ncommand with an additional \u0027--retired\u0027 parameter to\nlist only nodes which have the retired property.\n\nStory: #2005425\nTask: #38143\nDepends-On: https://review.opendev.org/703981\nChange-Id: I5fe379c4ff439b3a083ae819ce5b4bdbddd22b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/c5a7ec60c22c995893aa11bd05fd9609a85d9217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/c5a7ec60c22c995893aa11bd05fd9609a85d9217"}]},"branch":"refs/heads/master"},"53d5f4a86ed2cba7d1e9b4cb54b87534baab81c4":{"kind":"REWORK","_number":4,"created":"2020-01-28 09:34:54.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/82/703982/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-ironicclient","ref":"refs/changes/82/703982/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/4"}}},"commit":{"parents":[{"commit":"a64370701e8acff6aa28da87ccebd72664a48d34","subject":"Add allocation owner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/a64370701e8acff6aa28da87ccebd72664a48d34"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-17 12:51:37.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-28 09:34:19.000000000","tz":60},"subject":"Add support for retired{_reason} fields.","message":"Add support for retired{_reason} fields.\n\nThis change adds support to set and unset the \u0027retired\u0027\nand \u0027retired_reason\u0027 fields. It also extends the \u0027list\u0027\ncommand with an additional \u0027--retired\u0027 parameter to\nlist only nodes which have the retired property.\n\nStory: #2005425\nTask: #38143\nDepends-On: https://review.opendev.org/703981\nChange-Id: I5fe379c4ff439b3a083ae819ce5b4bdbddd22b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/53d5f4a86ed2cba7d1e9b4cb54b87534baab81c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/53d5f4a86ed2cba7d1e9b4cb54b87534baab81c4"}]},"branch":"refs/heads/master"},"81eebfc404396345055b50c124962c4069048481":{"kind":"REWORK","_number":5,"created":"2020-02-06 15:10:03.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/82/703982/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-ironicclient","ref":"refs/changes/82/703982/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-ironicclient refs/changes/82/703982/5"}}},"commit":{"parents":[{"commit":"a64370701e8acff6aa28da87ccebd72664a48d34","subject":"Add allocation owner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/a64370701e8acff6aa28da87ccebd72664a48d34"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-01-17 12:51:37.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-02-06 15:09:27.000000000","tz":60},"subject":"Add support for retired{_reason} fields.","message":"Add support for retired{_reason} fields.\n\nThis change adds support to set and unset the \u0027retired\u0027\nand \u0027retired_reason\u0027 fields. It also extends the \u0027list\u0027\ncommand with an additional \u0027--retired\u0027 parameter to\nlist only nodes which have the retired property.\n\nStory: #2005425\nTask: #38143\nDepends-On: https://review.opendev.org/703981\nChange-Id: I5fe379c4ff439b3a083ae819ce5b4bdbddd22b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/81eebfc404396345055b50c124962c4069048481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-ironicclient/commit/81eebfc404396345055b50c124962c4069048481"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
