)]}'
{"id":"openstack%2Fpython-keystoneclient~33661","triplet_id":"openstack%2Fpython-keystoneclient~master~Ice8724949a48bfad3b8b7c41b5f50a18a9ad9f42","project":"openstack/python-keystoneclient","branch":"master","topic":"bug/1175367","hashtags":[],"change_id":"Ice8724949a48bfad3b8b7c41b5f50a18a9ad9f42","subject":"Fix memcache encryption middleware","status":"MERGED","created":"2013-06-19 15:06:04.000000000","updated":"2013-06-19 17:44:20.000000000","submitted":"2013-06-19 17:44:20.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2c5dc29934310d6518909d5e86982a006c4ccf3","_number":33661,"virtual_id_number":33661,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-06-19 17:44:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":2807,"name":"Bryan D. Payne","email":"bdpayne@acm.org","username":"bdpayne"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-06-19 15:08:23.000000000","_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":2807,"name":"Bryan D. Payne","email":"bdpayne@acm.org","username":"bdpayne"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-06-19 15:08:23.000000000","_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"date":"2013-06-19 15:06:04.000000000","_account_id":2807,"name":"Bryan D. Payne","email":"bdpayne@acm.org","username":"bdpayne"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2807,"name":"Bryan D. Payne","email":"bdpayne@acm.org","username":"bdpayne"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-19 15:06:04.000000000","updated_by":{"_account_id":2807,"name":"Bryan D. Payne","email":"bdpayne@acm.org","username":"bdpayne"},"reviewer":{"_account_id":2807,"name":"Bryan D. Payne","email":"bdpayne@acm.org","username":"bdpayne"},"state":"REVIEWER"},{"updated":"2013-06-19 15:08:23.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"},{"updated":"2013-06-19 17:44:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"0f1c4d40b58fd799f4c5b2cd49ee80bf5a668f44","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-06-19 15:08:23.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\napproving based on reviews in bug 1175367","accounts_in_message":[],"_revision_number":1},{"id":"e78bed6a292de93adb41c689445cb9c12fbc1c5f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-19 15:08:25.000000000","message":"Patch Set 1:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"9aec593ddd450edd90ff3c0088fcc4132d0f4e9b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-19 15:58:43.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/33661/1/check/gate-python-keystoneclient-pep8/1230 : SUCCESS in 1m 06s\n- http://docs-draft.openstack.org/33661/1/check/gate-python-keystoneclient-docs/1081/doc/build/html/ : SUCCESS in 1m 31s\n- http://logs.openstack.org/33661/1/check/gate-python-keystoneclient-python26/1154 : SUCCESS in 1m 33s\n- http://logs.openstack.org/33661/1/check/gate-python-keystoneclient-python27/1157 : SUCCESS in 1m 13s\n- http://logs.openstack.org/33661/1/check/gate-tempest-devstack-vm-full/24620 : SUCCESS in 52m 04s\n- http://logs.openstack.org/33661/1/check/gate-tempest-devstack-vm-postgres-full/24326 : SUCCESS in 42m 49s\n- http://logs.openstack.org/33661/1/check/gate-tempest-devstack-vm-quantum/32105 : SUCCESS in 18m 58s\n- http://logs.openstack.org/33661/1/check/gate-grenade-devstack-vm/11452 : FAILURE in 39m 01s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"5f640eb4cfa66d463c5bfb4965e9fb9e4b783be3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-19 17:44:20.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1},{"id":"adbd521052c99f3bab565a921aa948865b4787cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-19 17:44:20.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/33661/1/gate/gate-python-keystoneclient-docs/1085/doc/build/html/ : SUCCESS in 1m 07s\n- http://logs.openstack.org/33661/1/gate/gate-python-keystoneclient-pep8/1234 : SUCCESS in 58s\n- http://logs.openstack.org/33661/1/gate/gate-python-keystoneclient-python26/1158 : SUCCESS in 2m 55s\n- http://logs.openstack.org/33661/1/gate/gate-python-keystoneclient-python27/1161 : SUCCESS in 1m 59s\n- http://logs.openstack.org/33661/1/gate/gate-tempest-devstack-vm-full/24664 : SUCCESS in 49m 55s\n- http://logs.openstack.org/33661/1/gate/gate-tempest-devstack-vm-postgres-full/24370 : SUCCESS in 58m 56s\n- http://logs.openstack.org/33661/1/gate/gate-tempest-devstack-vm-quantum/32154 : SUCCESS in 26m 52s\n- http://logs.openstack.org/33661/1/gate/gate-grenade-devstack-vm/11487 : FAILURE in 44m 57s (non-voting)\n","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eeefb784f24c37d5f56a421e1ccc911cace9385e","revisions":{"eeefb784f24c37d5f56a421e1ccc911cace9385e":{"kind":"REWORK","_number":1,"created":"2013-06-19 15:06:04.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/61/33661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-keystoneclient","ref":"refs/changes/61/33661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-keystoneclient refs/changes/61/33661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-keystoneclient refs/changes/61/33661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-keystoneclient refs/changes/61/33661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-keystoneclient refs/changes/61/33661/1"}}},"commit":{"parents":[{"commit":"1e3cf4bb2f1370d4117ce1e1d934838e1a475f32","subject":"Merge \"Use AuthRef for some client fields\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-keystoneclient/commit/1e3cf4bb2f1370d4117ce1e1d934838e1a475f32"}]}],"author":{"name":"Bryan D. Payne","email":"bdpayne@acm.org","date":"2013-06-07 16:34:25.000000000","tz":-420},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2013-06-19 15:05:02.000000000","tz":120},"subject":"Fix memcache encryption middleware","message":"Fix memcache encryption middleware\n\nThis fixes lp1175367 and lp1175368 by redesigning the memcache crypt\nmiddleware to not do dangerous things. It is forward compatible, but\nwill invalidate any existing ephemeral encrypted or signed memcache\nentries.\n\nChange-Id: Ice8724949a48bfad3b8b7c41b5f50a18a9ad9f42\nSigned-off-by: Bryan D. Payne \u003cbdpayne@acm.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-keystoneclient/commit/eeefb784f24c37d5f56a421e1ccc911cace9385e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-keystoneclient/commit/eeefb784f24c37d5f56a421e1ccc911cace9385e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
