)]}'
{"id":"openstack%2Fpython-manilaclient~1003456","triplet_id":"openstack%2Fpython-manilaclient~stable%2F2026.2~Iff10cd738fa5512e379bdb5c217055be5825d770","project":"openstack/python-manilaclient","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"Iff10cd738fa5512e379bdb5c217055be5825d770","subject":"Update .gitreview for stable/2026.2","status":"MERGED","created":"2026-09-02 10:54:40.000000000","updated":"2026-09-02 13:50:49.000000000","submitted":"2026-09-02 13:49:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003456-create-2026.2","meta_rev_id":"973e4e5d3e4c20e6dd4a847a6e6385c2f66b3712","_number":1003456,"virtual_id_number":1003456,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 13:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-09-02 13:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-09-02 13:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 11:09:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 11:52:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2026-09-02 13:29:15.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"ff5ce62fa6ae7351acdb546b77f65299e3284cea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-02 10:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a913d9fe03b209ab22aa62db277743c90bd9c47f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:09:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cc37940fb9e4e3d9c34ad71155da2ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ef2d67b4af4eb5a0b1037375f250b4 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b71bb533e9a4dc6b0b10418f2eb7ca2 : FAILURE in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61325971f2904fa7bf9b7f1d055d28e3 : SUCCESS in 4m 17s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/93248d01525d40e0b20cc3fb4c6003ba : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a71ac2cc2f34f6a7427bf084d2b60588102cd082","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-09-02 11:52:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60f5b10f5d2a320bdf52c43444b412f4d8db740e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:06:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1edf5a668f9a44b19671fc17701f3d22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5741a3896dfb4b65bc607fae610255d2 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e053f30420a24d8f94e1349f96cbfb5b : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cdef91559b24acb804708d976aeb46e : SUCCESS in 4m 15s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/d17e3ab7c83b4a0286328de9ba459fda : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a02446f8996fae3deb1fa68b86016a825e656b2d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-09-02 13:29:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a57a33deb7725f29694620268705f4d3332f8aab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:29:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1705615d984611478c4fd0d3b4bfa4f8c0b773c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:49:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d6d1d50ff5943379dfa3bc25b94b219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10ffc8f0c6049e5b4164287599f1e33 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f362f67e56684e7bba7fcf024ccc0928 : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1d01e9084ee4290b76633a9e2710622 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5ed5754c4581eb2c0cf250803a23c8a769bcacac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:49:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"973e4e5d3e4c20e6dd4a847a6e6385c2f66b3712","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:50:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369717dfcb9041c5a8af34b21ef121a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad0b230cdb94bb8bb655ec42a867ceb : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4752fa70a49152c8dd72a86f8e8ee4bdfe2ecee","revisions":{"c4752fa70a49152c8dd72a86f8e8ee4bdfe2ecee":{"kind":"REWORK","_number":1,"created":"2026-09-02 10:54:40.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/56/1003456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/56/1003456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/56/1003456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/56/1003456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/56/1003456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/56/1003456/1"}}},"commit":{"parents":[{"commit":"870fccd6b8a76844571d79581a1937d11ff13259","subject":"Add availability_zone filter to share list (2.97)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/870fccd6b8a76844571d79581a1937d11ff13259"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-02 10:54:39.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-02 10:54:39.000000000","tz":0},"subject":"Update .gitreview for stable/2026.2","message":"Update .gitreview for stable/2026.2\n\nChange-Id: Iff10cd738fa5512e379bdb5c217055be5825d770\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4752fa70a49152c8dd72a86f8e8ee4bdfe2ecee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4752fa70a49152c8dd72a86f8e8ee4bdfe2ecee"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
