)]}'
{"id":"openstack%2Fpython-manilaclient~1005162","triplet_id":"openstack%2Fpython-manilaclient~master~If4389c3ba90b3d68a67925d4663c7b9f27f867ee","project":"openstack/python-manilaclient","branch":"master","topic":"snapshot-followup","hashtags":[],"change_id":"If4389c3ba90b3d68a67925d4663c7b9f27f867ee","subject":"Fixed import styling issues and negative testing","status":"NEW","created":"2026-09-11 02:30:28.000000000","updated":"2026-09-11 03:47:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d011d683d784b94bdbeb469c75b23c5b9332d9f0","_number":1005162,"virtual_id_number":1005162,"owner":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 03:47:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 03:47:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1452e521babb5698aa11d958d8506e83c8c67547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-09-11 02:30:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d011d683d784b94bdbeb469c75b23c5b9332d9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 03:47:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f4a3fda642e42a09bddb1b2f3e7a035\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c91c7765f44d25b8dd3fbc04829a31 : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/218fd33d4816488ca89a608b46b62924 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc9c11626f444b5b985054522be56639 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f9115f3818f4fe288a3a4029383785d : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3bf27d9fb5154a4da6d1de1d79f7f16b : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00821d18c7e24df9922e2837261ca3b1 : SUCCESS in 5m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a19bf110fa042758e847f2041152b10 : SUCCESS in 4m 36s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/1772af5695ac4230a6d96493a6899737 : SUCCESS in 31m 19s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/28c0f75c36534555b872159a8bc811e9 : SUCCESS in 3m 36s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/69f65a77de3443d0897b8211cd275d14 : SUCCESS in 39m 35s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/a5e50a1369ba41c7ab2d4dcc44e46b37 : POST_FAILURE in 10m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/08f46e4491a344ac8365ab8d3d7217f1 : SUCCESS in 32m 44s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d5461278b34996ca4775c348a06189cf4b008d9","revisions":{"1d5461278b34996ca4775c348a06189cf4b008d9":{"kind":"REWORK","_number":1,"created":"2026-09-11 02:30:28.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/62/1005162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/62/1005162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/62/1005162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/62/1005162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/62/1005162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/62/1005162/1"}}},"commit":{"parents":[{"commit":"2bb39f33909bf99e4607e8f73775d1f03194ee64","subject":"Merge \"Add Functional test for share snapshot metadata \u0026 negative testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/2bb39f33909bf99e4607e8f73775d1f03194ee64"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-09-11 02:23:30.000000000","tz":-240},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-09-11 02:26:53.000000000","tz":-240},"subject":"Fixed import styling issues and negative testing","message":"Fixed import styling issues and negative testing\n\nChange-Id: If4389c3ba90b3d68a67925d4663c7b9f27f867ee\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/1d5461278b34996ca4775c348a06189cf4b008d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/1d5461278b34996ca4775c348a06189cf4b008d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
