)]}'
{"id":"openstack%2Fpython-manilaclient~878029","triplet_id":"openstack%2Fpython-manilaclient~stable%2Fxena~Ic654277feb002461cf7eae7966e2c1584495162a","project":"openstack/python-manilaclient","branch":"stable/xena","topic":"bug/198830","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-03-28 15:30:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic654277feb002461cf7eae7966e2c1584495162a","subject":"Make the OSC plugin handle the --insecure flag","status":"MERGED","created":"2023-03-21 17:08:35.000000000","updated":"2023-03-28 15:31:49.000000000","submitted":"2023-03-28 15:30:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878029-bug/198830","cherry_pick_of_change":878014,"cherry_pick_of_patch_set":1,"meta_rev_id":"d618dcea94d9c3ae848f088dbcd6fff355677e22","_number":878029,"virtual_id_number":878029,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-28 15:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-03-28 14:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-03-28 14:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":8104,"name":"June Yi","display_name":"gochist","email":"june.yi@samsung.com","username":"juneyi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 17:08:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":8104,"name":"June Yi","display_name":"gochist","email":"june.yi@samsung.com","username":"juneyi"},"state":"CC"},{"updated":"2023-03-21 18:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 14:21:04.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"46768d6c85661cb78950fa459fb739433a3394f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-21 17:08:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"e20643deccb2d8f80398f5a1554995aca95eba72","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-21 17:08:41.000000000","message":"Topic bug/198830-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"90249d71349f0a437d5f576e7da050c83b72aca4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-21 17:08:44.000000000","message":"Topic set to bug/198830","accounts_in_message":[],"_revision_number":1},{"id":"892f9f3607776dded424a4c125f287cd5e0a1202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 18:23:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1a9ca4b02c4dc598c2d6da70f6ef2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2266ad6950a940959a860d7c36a132d0 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e3e8a4bcbb7475883ea2d5a2e2a39d0 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ddb9198845644e7b495150b82eebef2 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15bef84f2749417db14655ad95e839e7 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29311cec138c474b8acde6bdaa50a7ee : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c5e30b9c7d468fb7922d7654f7106b : SUCCESS in 4m 34s (non-voting)\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/b2972597c8644e31b7c7b6491e8c6418 : SUCCESS in 1h 01m 09s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1d7aaede18ce4694807a9752bdfa2c1e : SUCCESS in 39m 13s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/b2dc6f1fe6a8466c855bcc53607d6e10 : SUCCESS in 1h 07m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ff8d6c065eba40bb96e60b102ee1a069 : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4201aef7d1507c75bd8a8af6f54aeb6e284b3f1a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-28 14:21:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68e60cbefbc274815f70e4165a7096846944dc8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:21:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30542c1647ace571b17efccbc8da694b50313a99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:30:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c521ae88d30b436c94430c4125a756b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71279b21a4949e48fc14bed6f8bd1a8 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc46f2d4ff2341af97446b667f3e89b5 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17de1df0ed244090a524368eec8c8d00 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f76806c480b4c4b9532d1fbc91017b2 : SUCCESS in 3m 44s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/b67ecad0d9fa44c79f5f07ee4ba468e9 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"476e430821b23446c21347b360f88cd48f003c02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:30:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d618dcea94d9c3ae848f088dbcd6fff355677e22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:31:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ada29e166ba944758373c86613eb1b8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f8577b0fd146e99ecceedac4a2ea06 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"046e92e08b1380049f0f0ebb9596b1632191eb3c","revisions":{"046e92e08b1380049f0f0ebb9596b1632191eb3c":{"kind":"REWORK","_number":1,"created":"2023-03-21 17:08:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/29/878029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/29/878029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/29/878029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/29/878029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/29/878029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/29/878029/1"}}},"commit":{"parents":[{"commit":"accd8e94f445c1c74b69ca2f18e06ec5dc085cbb","subject":"Add FIPS testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/accd8e94f445c1c74b69ca2f18e06ec5dc085cbb"}]}],"author":{"name":"June Yi","email":"june.yi@samsung.com","date":"2022-08-31 15:24:56.000000000","tz":540},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-03-21 17:08:35.000000000","tz":0},"subject":"Make the OSC plugin handle the --insecure flag","message":"Make the OSC plugin handle the --insecure flag\n\nCloses-Bug: #1988304\nChange-Id: Ic654277feb002461cf7eae7966e2c1584495162a\nSigned-off-by: June Yi \u003cjune.yi@samsung.com\u003e\n(cherry picked from commit 67ebf68079c86341da7ddd7a8c0b1cc654be70cf)\n(cherry picked from commit 29945d3a043e95a41b24c7dee344f6296832afb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/046e92e08b1380049f0f0ebb9596b1632191eb3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/046e92e08b1380049f0f0ebb9596b1632191eb3c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
