)]}'
{"id":"openstack%2Fpython-manilaclient~967138","triplet_id":"openstack%2Fpython-manilaclient~stable%2F2025.2~I2d971f2c4771d62a1776c995ef8a16594fd04c99","project":"openstack/python-manilaclient","branch":"stable/2025.2","topic":"bug/2057951","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-07-30 22:00:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d971f2c4771d62a1776c995ef8a16594fd04c99","subject":"fix respect region name in client initialization","status":"MERGED","created":"2025-11-13 22:23:17.000000000","updated":"2026-07-30 22:01:11.000000000","submitted":"2026-07-30 22:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967138-bug/2057951","cherry_pick_of_change":964719,"cherry_pick_of_patch_set":4,"meta_rev_id":"b5082ce16573fb5200b23b903aab05af381a087c","_number":967138,"virtual_id_number":967138,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-30 22:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-02-26 22:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2026-02-26 22:00:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 22:23:17.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":38471,"name":"Denver Baraka","display_name":"Denver Baraka","email":"denverbaraka@gmail.com","username":"denver-baraka"},"state":"CC"},{"updated":"2025-11-13 23:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 22:00:49.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"a14223505fb84d093ddf624d440d852aedd880c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-13 22:23:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d1489ddd7f54be225cdf29d25fa123e25c7a4699","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-13 22:23:42.000000000","message":"Topic bug/2057951-stable/2025.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"b83592c9fe90bb9058e0239335abeb03d2655116","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-11-13 22:23:45.000000000","message":"Topic set to bug/2057951","accounts_in_message":[],"_revision_number":1},{"id":"d247c161978e260bb376d0071a86ba1090a74319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 23:19:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd305158271428684d0b40714440097\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7fca1758eba4dcc87b3ceb50f9a8c5f : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03807ac84ff04f4692a319481b6d2611 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0f555c05a0489385c929481d6cad85 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e82ec8bf80d4cbbbbed16d45cad4699 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6fac6270e374075a57ac63c15593f96 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a7ad4523fa84d3898aa1abe46761d39 : SUCCESS in 8m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9617c09957064863bc266b648ab18f11 : SUCCESS in 3m 18s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/4c7a982458a24b96896ebf1e58c306a3 : SUCCESS in 55m 32s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/10dbbd68385d41adba67dd68a53a7c2b : SUCCESS in 2m 12s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/090e8a7714aa4c22a808f75ecebe25dd : FAILURE in 12m 24s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/077ac4136ea1497d89aa54f9d95aa1c4 : POST_FAILURE in 7m 54s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/488747b4bc3b44dc948d92a02c1c5d22 : FAILURE in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15def4b34f48b2eb464255ee97d5a345d9c4490b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-26 22:00:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc62f33975766376ee447862c472c889bf4fe5ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 22:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68babde989a074eba9585c702a52024365b7a64e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 22:45:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e51d084922784828a1e22e0412bd97d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088fc33364624f5dbc24d67fc7c74af8 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be51c6a5e3b4b41b3f3b4ccd102220c : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d42294463cbc459ea9464a815f223c53 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/999d08add5eb4261a63485b3618052d3 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ea009ed38e44949fad436af2c7399a : SUCCESS in 2m 55s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/0d0f00a4274e49b0924e9e55bd7503a4 : FAILURE in 20m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5dab9209429466e2e0418a880a22ca2c9de2efcc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-19 05:47:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f2230a5a604bfb2bbdec17e4b50f47afdc0735b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 06:06:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4521521915b048f4a8638578bdbfec8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9360c3d73a6643db9adac2f745e9144e : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d04cd40b15240adb9691a0c90f6f19a : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49bc7bf210484ed0bd697f98b0b53a70 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9b3c86e2af54977a5de79cdef2a3098 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e540967246641968f5254481c54b867 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ded302190ca4d32ac7285700641c67b : SUCCESS in 4m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6ae6f59f524db8973fbc25643add9d : SUCCESS in 2m 40s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/e879856925d747668379c3b9a16264de : FAILURE in 10m 24s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/38d9b083d7db460992e71d3ceca4f0d2 : SUCCESS in 2m 32s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0011c510a6b64d7fbd9fd7f04378a55d : SUCCESS in 18m 10s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/f078dd300aff4855b59d443f6099f90c : POST_FAILURE in 9m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c9cab5f0ee294d51a6b4d889b5a18976 : SUCCESS in 16m 16s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c3f4e14d04eb2ed1671ac9ce1ff6b8c0596d112c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-30 20:10:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95767eebb039757d36ba2fde648518b26f83c31a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 21:00:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c86875a5c8784b0eaa72d89b806389e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79b3aa20e6174a59b23fb450831cea18 : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/185140757f5241fa8ac4efe4615dca58 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874f9c4ea461456fbeb40f5dde4aae7d : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1172f2a506da428396455d711644ef53 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce6a5c9b8ffb4e0ba0852a8bcd92ce1b : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c08fec8b6d84e59b5f40a5cd573a663 : SUCCESS in 3m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1391bd22489b4f5a9bfe40ca728600b8 : SUCCESS in 4m 02s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/7f18e49ed4c14e8aa68120f8a0dc8c90 : SUCCESS in 42m 13s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/5db06b502c6e41e1b7c69271c5a01b00 : SUCCESS in 1m 55s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/59d44135bd5b4bc98fbc2187c9bb59f2 : SUCCESS in 15m 56s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/68206beac9bd46f882f8e560bdfc2b85 : POST_FAILURE in 9m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c42e1268b32142e289fdd0473f355998 : SUCCESS in 33m 26s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"3252d92203ac3dd2a35d356a1d299a6984dc489a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 21:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a797cca73803ed3446e70a15421e899fea07cbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 22:00:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d20208eae05f42e9ad1627a1903d3c34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c1efa0fe1740c1947784d204d0787a : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdbf0820d396409d83334c0861d3275b : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07bddb06b0824296ab772fdc120fd114 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ad07f9454af4a4485153bfa5d67f17b : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f162e9fb85945a8987af9c866bb7985 : SUCCESS in 4m 44s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/d7b62ab1dfb149319413d1b279fc2f69 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"be5931a28d27136385eae266cbaadbc8908dcd06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 22:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b5082ce16573fb5200b23b903aab05af381a087c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 22:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c50843dc2234e59b25cacb1e8a1330a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a98e769a0e461e87f90d437151b502 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0ac36d606bb448a97684ef4b204b4a8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4b7f1619ecc4e31711bfea929813461d1b025c7","revisions":{"c4b7f1619ecc4e31711bfea929813461d1b025c7":{"kind":"REWORK","_number":1,"created":"2025-11-13 22:23:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/967138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/38/967138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/38/967138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/38/967138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/38/967138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/38/967138/1"}}},"commit":{"parents":[{"commit":"b44aae26b80e21db3f3c803ea5259fa7f58573e4","subject":"Fixes an issue that prevented ensure shares to be run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/b44aae26b80e21db3f3c803ea5259fa7f58573e4"}]}],"author":{"name":"denver-baraka","email":"denverbaraka@gmail.com","date":"2025-10-23 17:26:59.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-11-13 22:23:17.000000000","tz":0},"subject":"fix respect region name in client initialization","message":"fix respect region name in client initialization\n\nwhen initializing manila python client, the region\nname was being ignored. As a result, the client\ncould connect to the wrong region endpoint.\n\nThis change ensures that region name is passed to\nsession.get_endpoint() so the correct endpoint is\nused.\n\nTest,\n-Updated unit tests to verify that the region name\n is respected during client initialization.\n\nCloses-Bug: #2057951\nChange-Id: I2d971f2c4771d62a1776c995ef8a16594fd04c99\nSigned-off-by: denver-baraka \u003cdenverbaraka@gmail.com\u003e\nAssisted-By: Copilot\n(cherry picked from commit 8019cc5ac0850e00ff8d52a82a627aa4a97ab11c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b7f1619ecc4e31711bfea929813461d1b025c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b7f1619ecc4e31711bfea929813461d1b025c7"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
