)]}'
{"id":"openstack%2Fpython-manilaclient~967758","triplet_id":"openstack%2Fpython-manilaclient~master~I43d95711c836e1f64dbd8b325c0a2686a33e51bf","project":"openstack/python-manilaclient","branch":"master","topic":"share_snapshots","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-09-02 19:29:17.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-02 18:57:26.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"37481":{"account":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"last_update":"2026-09-02 19:29:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43d95711c836e1f64dbd8b325c0a2686a33e51bf","subject":"Add Functional test for share snapshot metadata \u0026 negative testing","status":"MERGED","created":"2025-11-19 20:19:38.000000000","updated":"2026-09-02 19:30:35.000000000","submitted":"2026-09-02 19:29:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"967758-share_snapshots","meta_rev_id":"140e770bcd7432ba7b31d3cb265861f4bb4a1676","_number":967758,"virtual_id_number":967758,"owner":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 19:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2026-09-02 18:57:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-09-02 18:57:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 21:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 22:21:55.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-25 19:44:02.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2026-07-29 16:26:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"42a9b10aa21b92367f9fb15e49499ac65fdc28be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2025-11-19 20:19:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2fa37c2b000f89a3f0638046a61b92c50ffbcae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 21:22:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6692efe17ee4ccfbf7cc93e8139d7b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0645b8a4ccb4fd9b4a3918fb2adb353 : SUCCESS in 7m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3ab182d0b841beaf099d09c86e43a5 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8552b34420014891ac3b9ec3db2c141d : FAILURE in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8fb77dc8a8c4994b6e131e64080e6aa : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f725d37ff704feaab709b6a30ead0de : SUCCESS in 6m 39s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b76eb9b779ab4e07b95f2dc766d31626 : SUCCESS in 5m 30s (non-voting)\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/1b3d0421c35b40cf92551ad74617f6a8 : FAILURE in 51m 33s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/f9d7db2941074cf7af4a166d1283013b : SUCCESS in 2m 50s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7595afcebbc04ef1b1d0782f1b69007a : FAILURE in 11m 31s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/0a2ca7db635b4e8da184b75b3980fe3e : POST_FAILURE in 9m 18s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a45732a75e2e4feca3881e8241c9cba4 : FAILURE in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9579758234baae2ef867021ea2a722785e71e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2025-12-17 20:27:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"80251645a2811013ae55525f434535f36d611cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 21:28:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af8d963432574ee088f38c3b5253e416\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205563f6f29c441381a2a29c5e092ec3 : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f641ab18d074257a98a5b1c96b81f35 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b638bb16223049b3ae98188804306dad : FAILURE in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b562205eb3e4e4a9b33ad3ffdfa9e03 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/918a7e5063504ff2870a251bb3054efe : SUCCESS in 6m 36s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7b3c7126927f4476ab8ba7446ad0ef89 : SUCCESS in 2m 39s (non-voting)\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/04fddf6e6bae47ba9bdd4f40fa0f5394 : FAILURE in 56m 04s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/8129d800b87b4e5db783226d58de0fc5 : SUCCESS in 2m 55s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/22099c671d10404eae7c5498c7d5b0bb : SUCCESS in 18m 43s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/c3799617ae77403483c3d53cdce1ccf5 : POST_FAILURE in 20m 56s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/30eb9d5e421b4391957f41305a7c88ba : SUCCESS in 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc86e65e965739e094a30f5e47e7daa355e9f280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-01-27 01:50:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4806e5cb09d1f063dd846b5d4c39851ae754a73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 02:45:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2a743df9f794713aa9817ec83860648\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245753a4cde94829b1e058cb960bae0e : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8efebc8a53d4a02a2c21076f86bc741 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd75c2cfdcb4ea2950e0cf7d8338aae : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f18281f513ca422c968a6367a972a7be : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/392d528e49664d5c9481a6e5b4a4cbdb : SUCCESS in 2m 14s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/3dcaf0659c334a59bf52974d5c5b1203 : FAILURE in 53m 00s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/e01e928d9d844ea5937e6fe38a87a565 : SUCCESS in 3m 03s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6ecb825685164142b20e71fa0e25cfb6 : SUCCESS in 16m 59s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/55974cf6e6de40ac84123ea99d9e2c37 : POST_FAILURE in 18m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4929c6403f08476eacab9a7f16594406 : SUCCESS in 35m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f7a7f8f61fca8a66ca44ff78306e10d519f5370b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-27 22:21:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5ce44a6dcf6f91e77e4491a0a26e8d62d8776b87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-01-28 19:12:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"76e36d20e462484ebf9fa7b2524febabafe0cfd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 20:52:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e227af23ef7c4eaa8caff707e02a64bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4617c620f5ed4dd0acb5bb64d79653de : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26fd94dccd9e487691644bf0000a5bf4 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e802d797c3e45619e30091d0e91bc7b : FAILURE in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e0365cab91c47da95b713c6f50d3d22 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f50137476df343c4b45fddf97a9a97c4 : SUCCESS in 3m 37s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/eb499720a9d2424e9169fc7b8d31d6ab : FAILURE in 1h 00m 24s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/c427b477e1d740bd8c2513d1f303e9c9 : SUCCESS in 3m 26s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b129dfc6817b4b3fb443d6a8c214c8e5 : SUCCESS in 36m 22s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/b7298762d92845c4b3ccb56fdc7cd3aa : POST_FAILURE in 17m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/031dc2f26af84d3e95b3685a73a35e26 : SUCCESS in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1f64d22543c0e5c6fa82e30f6302ae9887688341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-02-11 19:51:44.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cccd03585ecb5427813b412bc231d4dcc2b61b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 20:29:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63ec02f9e13c4021b153b6edeea63528\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd86629887c64383a4c1325c6d8385f8 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d441692b41414c909580b64ee409bef8 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bf8bebfec124a97bdde53e2429166ed : FAILURE in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2a6355a8a2140e496f3f271f7343716 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f9f84334cba4b4eb826d8eeebbcfdbd : SUCCESS in 1m 56s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/fb9bc29dfd40455eb5793a1692d8161f : FAILURE in 21m 48s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/30d85abc33d74dc4971ce3ef0c628d8f : SUCCESS in 3m 01s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7bff95587cf04c1facfe7df48ff6c38a : SUCCESS in 32m 59s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/abcdbfa4b14b40f595c9b6fbc167f7ce : POST_FAILURE in 11m 11s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2ed3017cc9604125a26ac856bf504f2a : SUCCESS in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9a6e64495172f7eac68c0addc9f224a0e1a30af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-02-24 03:32:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bf706912778da4218bf33677186fd5fd6a156f67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 04:07:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80695d66d012495c9d02025cc7f4036e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed02b2f5246e49feb207df985e3c4716 : SUCCESS in 7m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1877bc1dca484c58a9cde5de43e99abc : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdd605251cdb4550a29b6b35cd455326 : FAILURE in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfad74e738c34fc7baf7770268893085 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d43365df5c334f0b87974170c415d3fd : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26da7fdfee664a65aaccf11073545912 : SUCCESS in 4m 12s (non-voting)\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/caaf5080450540d7899488056db32f96 : FAILURE in 28m 56s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/937d86a6e2f1430b9da9079408ff0295 : SUCCESS in 2m 21s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3b0989a29c7e4491ad4f3aa18abb22c4 : FAILURE in 20m 38s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/a1779fe64eeb4d67a91209856df2c6c9 : POST_FAILURE in 30m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0080f3265dbf4fcc9530367596d9a3da : FAILURE in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e61a8b28ffcc8060165b41fd487e2b2faaea197d","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-02-25 17:32:42.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"26d7aca0beac2291c98fc186e985b18b4eb38b55","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-25 19:44:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dfc858e8fe997c185eca59d266195769b4f26fff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-02-27 00:57:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"afbdce9ca77d3c567078033bde8bad85925fd6b0","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-02-27 01:00:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"05d85e34a9553ec9303fe249d1506ae0368f663a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 01:34:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1deb556a9c15403bb71bc4e6caf08915\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69fc9ebf43f54a7e9b86fe03c069f11f : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d893da1cb0c4fd6a57606014480e1a7 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d23d8791fb4737a1a04be59665fde6 : FAILURE in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e75ba78921fe470a858c097788f4426a : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/461d5e833ca44ff09c8053e322a9819e : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a2df43baf834544b7ce95fe77beee52 : SUCCESS in 7m 42s (non-voting)\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/863c4c678eaa415f97b66b6dd2b7eed8 : SUCCESS in 31m 46s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/643380e6cba147d0911dc517184a2985 : SUCCESS in 4m 06s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/194135658d0843f190c97d95d3f383a5 : FAILURE in 12m 44s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/a90a82740d6e41cba422ecb42983a49c : POST_FAILURE in 16m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2fc2acd2fa134b3db38bf63b48bdddbf : FAILURE in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1a16b1a7cebd8e6d80df4c1b249c993e20ef7642","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-15 18:51:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a48a2242131c1a2f91d606a49dd8a453da78da48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 19:34:02.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12a0fe195e474d6db214d2283683cc17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed606bbba0aa4b36a626904eb59bdb86 : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3398da99a25340dc83328ada7685aaee : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29422f8f2c7a4dcd83b7486fcafa86e2 : FAILURE in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ab58bffcb5044b4b26f9c554d5939a0 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b68b35cd9fd415790ecd1dd0eaa61bc : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/789b450f6a144f76be7c0451b49dcc6b : SUCCESS in 6m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb09901d38a746e3834689fa6efaee3b : SUCCESS in 5m 00s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/7b654639eb6446448c7429ee4faa5271 : SUCCESS in 37m 07s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/f9f9efa886104f9d8c8737c7e55f40ec : SUCCESS in 3m 29s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4fc0ba8860534d1fafc639820b166e11 : SUCCESS in 36m 59s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/b289558f80994e7cb85d22f379e43de5 : POST_FAILURE in 14m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1c2cf36446b1490fab8216fcdf9284ca : SUCCESS in 24m 19s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"80984526d6aa8e11e8171142119ea5b1e231c9c5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-29 16:26:54.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5a9eb9e265f2dd4ec1960976b8861cdd4915b598","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-08-25 18:40:46.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"ebfe36e8fd934b9d2e028fe7c30aa1686b4f55e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 19:25:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75f4280bc5d24ae3b59651dcd1b2307f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ddffeb6312e46399b78c8cd316d42bd : SUCCESS in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d061f67f78024e5aa1104b210aaf8471 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897040fc8b7d41e59670deaf282d19b7 : FAILURE in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da1d77787503414f90079944d67f2566 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/415ff59b6c794faeb842141820188017 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/359d426492f84373a2a45a83c9194aaa : SUCCESS in 6m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13ba2bfb301149c28ab6f095029ae591 : SUCCESS in 4m 22s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/1cf06fb496bd49bfb261d3b73ba68325 : POST_FAILURE in 15m 51s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/bb6f868a233343e59d5d461499ce7476 : SUCCESS in 3m 41s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b84b479d8037488da42686245abc2140 : SUCCESS in 34m 17s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/448e0169bf8b40b38936eb5d2f30fab4 : POST_FAILURE in 18m 40s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/288b4327ff9c46cdbb6d1bcf579a90dc : SUCCESS in 26m 59s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"9019bae17dfdb4a8bd348cad664418fbbdca6684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-08-26 01:59:20.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c1c6800a269505c3b73771478a317452c7b506eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:38:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb97e75f94b74326b2c412a8180b188b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27494bdccfe4c08ae9af68892651e1f : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c8313df283e4ddf84d58809482ea13d : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f07dd747ae4a988405dc452a6a9ad4 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9d7595de694421595d74e945b200dcd : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96c563ce060a43eeac2116b524f546f2 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83b2963a8c24427dbd1e7c37e14c99f9 : SUCCESS in 3m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4f12fc57f9d455c84f3e201cc967ff7 : SUCCESS in 2m 36s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/a344b9a0bb56401287e5628b25d3faae : FAILURE in 27m 58s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/bf30009bf0144fff867d53209e5e999a : SUCCESS in 2m 39s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a62e8c152801434cb4c936c5f4cbb19c : SUCCESS in 16m 22s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/9e646aea789d424d98be616db04e8a6e : POST_FAILURE in 12m 49s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/96ddc1de27024bada52cb3f654a0bb60 : SUCCESS in 32m 24s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"9567db9c65793421ca57c5b97d138957494a79a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-08-26 02:41:02.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"a2065beb332c6f31ea4d361ee2e43c76b926accd","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-08-26 02:45:33.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1e50dffbd4dd498e577dbff919ee3e3cd8371690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 04:09:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62f6ba558f4a45bfbefe692474d25e2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a718f73922c64efa876aef28f9818d76 : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71e98163f21b4de88385c9011e33205b : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1396191ffaa48ffbc0b0f77bbf3d8c9 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b3bc09dfb6a429090c28ca2769ca491 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43e89e70758345a69e7f8f45fee97be0 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18f337b88dd441918bf80da74ff9fee6 : SUCCESS in 8m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99fdef9456744d4bb5274c95d703df47 : SUCCESS in 2m 52s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/a61be461c4784f82bcd53635ae241d2b : FAILURE in 34m 13s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/16b3d9510fdb416ea6402b8a2a75fc94 : SUCCESS in 3m 42s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/093b0ef48de34125b18c096737e9e2cb : SUCCESS in 34m 31s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/3fa9ac2f8aab466eaa216f5c413ce10b : POST_FAILURE in 22m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c7400852695f412c9302d267eaa36e41 : SUCCESS in 16m 01s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"97669eb7b3b90d9fb66ad154ac89f61fb520ba5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"date":"2026-09-02 03:24:06.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"b64d9f4f364b50f39444d3375e36365f9ce1348a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 03:57:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/119ee9d96cb34096b63f8a31f00a0fc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef39054a49b94fc68490b4aded3d3748 : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1534c0463acc47c3994ebec8b20dc6cf : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532238a1ea39477aa73680a64555d739 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28ddfa4f566648418ad371a28b4f0ef4 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b09361724d7b488dbfffe350d648cfad : SUCCESS in 2m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4b4b955b53b45dc9606bd5b839ceab4 : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6039b095db944a8e931e68920ccf2da3 : SUCCESS in 2m 41s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/84d331bdb38445ea885c47bbcfc805c3 : SUCCESS in 24m 02s\n- python-manilaclient-tox-bandit https://zuul.opendev.org/t/openstack/build/dcfdb3e14e3a4b1285d6884311192324 : SUCCESS in 3m 19s\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c26804dca250417383241517599b43f9 : SUCCESS in 15m 05s (non-voting)\n- python-manilaclient-functional-fips https://zuul.opendev.org/t/openstack/build/ab5967ff006c4b62a989a91d8e6267fa : POST_FAILURE in 18m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/70f01a56a7f64ebab0f507bcd48ae923 : SUCCESS in 31m 16s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-manilaclient-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"b6b2d4523def4ba71d28d2440364863c07ae145f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-02 18:57:26.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"815494772b3ac94544bcf8f7cf14a84ad9a1e435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 18:58:43.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"40f5dac12fab77be68065d6d50228c91389e11c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:29:17.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fb4a86d3d274f33aaecf7a26a596823\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761ee78be6654d8c9848cea82ce07a84 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5aea82f3ae148d6b5d60066778f0c7c : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00e089e845384f168cc059ff4d2adfc6 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb706ae421d745788d29d776db3eedf7 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898f39c041224b9abc5f1969bdcb1e4d : SUCCESS in 3m 00s\n- python-manilaclient-functional https://zuul.opendev.org/t/openstack/build/d79a58c02c864db98967162bb773c56e : SUCCESS in 26m 39s","accounts_in_message":[],"_revision_number":11},{"id":"9f3193661131f3f09f1984815256327610b77ac7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:29:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"140e770bcd7432ba7b31d3cb265861f4bb4a1676","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:30:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db6f7e8c7abf4599bbf0e7a6c9651b1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153c627b793543a2afa2b67514f5b7f5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"140eebc99d9dbbdf0b5f5151090e0179f585a96a","revisions":{"ab62e6b3afcb286eeacd8665164760f5f740a3be":{"kind":"REWORK","_number":1,"created":"2025-11-19 20:19:38.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/1"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 20:13:33.000000000","tz":-300},"subject":"Add functional test for share snapshot metadata","message":"Add functional test for share snapshot metadata\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/ab62e6b3afcb286eeacd8665164760f5f740a3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/ab62e6b3afcb286eeacd8665164760f5f740a3be"}]},"branch":"refs/heads/master"},"372a77bbcb25f1f362b6a4a990cea8abbd7458a3":{"kind":"REWORK","_number":2,"created":"2025-12-17 20:27:18.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/2"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-12-17 20:26:37.000000000","tz":-300},"subject":"Add functional test for share snapshot metadata","message":"Add functional test for share snapshot metadata\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/372a77bbcb25f1f362b6a4a990cea8abbd7458a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/372a77bbcb25f1f362b6a4a990cea8abbd7458a3"}]},"branch":"refs/heads/master"},"02c0f6328c2de7d94d8c0afb1b77520ebfef9fe4":{"kind":"REWORK","_number":3,"created":"2026-01-27 01:50:31.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/3"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-01-27 01:49:31.000000000","tz":-300},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/02c0f6328c2de7d94d8c0afb1b77520ebfef9fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/02c0f6328c2de7d94d8c0afb1b77520ebfef9fe4"}]},"branch":"refs/heads/master"},"c302aba50c54f3361f79e627559ea366b99e3fb0":{"kind":"REWORK","_number":4,"created":"2026-01-28 19:12:56.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/4"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-01-28 19:12:31.000000000","tz":-300},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c302aba50c54f3361f79e627559ea366b99e3fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c302aba50c54f3361f79e627559ea366b99e3fb0"}]},"branch":"refs/heads/master"},"9e1795d66ccc610e21e93939eac6a3a4ec40c862":{"kind":"REWORK","_number":5,"created":"2026-02-11 19:51:44.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/5"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-02-11 19:51:11.000000000","tz":-300},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/9e1795d66ccc610e21e93939eac6a3a4ec40c862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/9e1795d66ccc610e21e93939eac6a3a4ec40c862"}]},"branch":"refs/heads/master"},"36b73b74fa5205f55974a9ddeb735aa7c8bf3a28":{"kind":"REWORK","_number":6,"created":"2026-02-24 03:32:11.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/6"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-02-24 03:31:48.000000000","tz":-300},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/36b73b74fa5205f55974a9ddeb735aa7c8bf3a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/36b73b74fa5205f55974a9ddeb735aa7c8bf3a28"}]},"branch":"refs/heads/master"},"cfe6a802110244a6d4c503908de0c8ea62fb9f7d":{"kind":"REWORK","_number":7,"created":"2026-02-27 00:57:45.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/7"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-02-27 00:57:26.000000000","tz":-300},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/cfe6a802110244a6d4c503908de0c8ea62fb9f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/cfe6a802110244a6d4c503908de0c8ea62fb9f7d"}]},"branch":"refs/heads/master"},"96a17ad278d39cca9183d4fdff474907e5009220":{"kind":"REWORK","_number":8,"created":"2026-08-25 18:40:46.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/8"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-08-25 18:40:20.000000000","tz":-240},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/96a17ad278d39cca9183d4fdff474907e5009220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/96a17ad278d39cca9183d4fdff474907e5009220"}]},"branch":"refs/heads/master"},"a7d4ee32191a4a6629b97774b87417927a79088b":{"kind":"REWORK","_number":9,"created":"2026-08-26 01:59:20.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/9"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-08-26 01:59:03.000000000","tz":-240},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/a7d4ee32191a4a6629b97774b87417927a79088b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/a7d4ee32191a4a6629b97774b87417927a79088b"}]},"branch":"refs/heads/master"},"36ca24fcda6ef6d27abde766a287e2c426590909":{"kind":"REWORK","_number":10,"created":"2026-08-26 02:41:02.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/10"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-08-26 02:39:40.000000000","tz":-240},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/36ca24fcda6ef6d27abde766a287e2c426590909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/36ca24fcda6ef6d27abde766a287e2c426590909"}]},"branch":"refs/heads/master"},"140eebc99d9dbbdf0b5f5151090e0179f585a96a":{"kind":"REWORK","_number":11,"created":"2026-09-02 03:24:06.000000000","uploader":{"_account_id":37481,"name":"Maria Gallego","display_name":"Maria Isabel","email":"gallegomaria0204@gmail.com","username":"marissa0204"},"ref":"refs/changes/58/967758/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-manilaclient","ref":"refs/changes/58/967758/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-manilaclient refs/changes/58/967758/11"}}},"commit":{"parents":[{"commit":"c4b53ce179fcf003aae914b90b0501eae26e3524","subject":"Add a git-blame-ignore-revs file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/c4b53ce179fcf003aae914b90b0501eae26e3524"}]}],"author":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2025-11-19 19:56:00.000000000","tz":-300},"committer":{"name":"marissa0204","email":"gallegomaria0204@gmail.com","date":"2026-09-02 03:23:21.000000000","tz":-240},"subject":"Add Functional test for share snapshot metadata \u0026 negative testing","message":"Add Functional test for share snapshot metadata \u0026 negative testing\n\nChange-Id: I43d95711c836e1f64dbd8b325c0a2686a33e51bf\nSigned-off-by: marissa0204 \u003cgallegomaria0204@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/140eebc99d9dbbdf0b5f5151090e0179f585a96a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-manilaclient/commit/140eebc99d9dbbdf0b5f5151090e0179f585a96a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
